)]}'
{"id":"openstack%2Freleases~801167","triplet_id":"openstack%2Freleases~master~I0af4b15e95d2b2cbcec01bd3f99ca578d0006740","project":"openstack/releases","branch":"master","topic":"train-last","hashtags":[],"change_id":"I0af4b15e95d2b2cbcec01bd3f99ca578d0006740","subject":"[barbican-tempest-plugin] Tag train-last","status":"MERGED","created":"2021-07-17 00:06:39.000000000","updated":"2021-08-23 20:41:50.000000000","submitted":"2021-08-23 20:41:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"801167-train-last","meta_rev_id":"c08cb85508499d0526a5588b8345566233a75360","_number":801167,"virtual_id_number":801167,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 20:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-23 20:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-10 13:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-08-19 14:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-08-23 20:28:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-10 13:18:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"}],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-17 00:07:12.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2021-07-17 00:07:12.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-07-17 00:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-17 00:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 15:00:17.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-08-16 08:22:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-23 20:28:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"f51cdcb67b7975cf91c958e0beec0ad94712bbac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-17 00:06:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cf3deb7ba1645b1009eb890d23a8cdb17c48448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 00:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a60f3b08794950afce17c651bfda7f : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86887e526c4e40b3884722f4e5e5b218 : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f64edfc831324b34b1b18791420f2f9b : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"99103315432e5cc76d92f48be0f88549bf5ee43f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-08-10 13:17:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"34f698d62bdbb0f04983aaefc6a268dc466b8ed3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:18:54.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"4e82d0040668c29e95f51c40c5903135fb735692","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-08-10 15:00:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3835f42f27864e679bf08699c93547592eac521c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 08:22:12.000000000","message":"Patch Set 1: Code-Review-1\n\n+1 to highlight Luigi\u0027s comment.","accounts_in_message":[],"_revision_number":1},{"id":"c952949f99c7edb2b475f2517dacd5f58d35f6d9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-16 14:29:46.000000000","message":"Patch Set 1:\n\nI do think we need to update it to latest commit. Idea of this tag is to have a compatible tag with Tempest train-last.\n\nI picked the latest commit of the time Tempest created tag https://review.opendev.org/c/openstack/releases/+/801133.\n\nNot sure if any compatible issue while picking 798008 but that is not needed as such. current has is tested compatible tag with train-last of Tempest","accounts_in_message":[],"_revision_number":1},{"id":"2aaeccaffe9d2261b683b1a90e3b7792356c0f39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-17 07:04:44.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"9d6b5cff47ae87db4801e624752caef0b1d696c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 14:00:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9eac9724d221ae8074314a3ac82bdf6309719fca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 20:28:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e087638e9b95ad3e812bf2774fea9256da94801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 20:28:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e267f9771cba63f1155cfc5e8c5905910a1da563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 20:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779a26d5f5854f4e80bc284c03f40898 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c780188cd4a45f2b341164ed0b7a0de : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4a428445de04740b08bcbc5dc719284 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c08cb85508499d0526a5588b8345566233a75360","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 20:41:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2ac58fb1fe4d6c0c26a8ace9ab6f6d912a64839","revisions":{"e2ac58fb1fe4d6c0c26a8ace9ab6f6d912a64839":{"kind":"REWORK","_number":1,"created":"2021-07-17 00:06:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/801167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/801167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/801167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/801167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/801167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/801167/1"}}},"commit":{"parents":[{"commit":"3da0153e40e380045db115c8ae5e5b24fcf6cdf9","subject":"Merge \"Propose to retire monasca-analytics from deliverables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3da0153e40e380045db115c8ae5e5b24fcf6cdf9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-17 00:05:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-17 00:05:58.000000000","tz":-300},"subject":"[barbican-tempest-plugin] Tag train-last","message":"[barbican-tempest-plugin] Tag train-last\n\nTrain branch is in Extended Maintenance now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins is used to test the supported stable\nbranches.\n\nOnce any stable branch is moved to EM state then, Tempest and\nits plugins compatible tag[2] needs to be released so that we can\nkeep testing the EM stable branches with that tag once master\nTempest and its plugins are not compatible[3].\n\nTagging \u0027train-last\u0027[4] as well as a new version also with same hash.\n\n[1] https://releases.openstack.org/train/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n[4] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: I0af4b15e95d2b2cbcec01bd3f99ca578d0006740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2ac58fb1fe4d6c0c26a8ace9ab6f6d912a64839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2ac58fb1fe4d6c0c26a8ace9ab6f6d912a64839"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
