)]}'
{"id":"openstack%2Freleases~801353","triplet_id":"openstack%2Freleases~master~I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de","subject":"Makes victoria tripleo release","status":"MERGED","created":"2021-07-19 16:28:34.000000000","updated":"2021-07-22 10:33:14.000000000","submitted":"2021-07-22 10:33:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"801353","meta_rev_id":"b2b7b069dbb6963a76d4b6b546068f1486336f9b","_number":801353,"virtual_id_number":801353,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 10:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-07-22 10:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-07-21 14:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-07-22 10:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-07-21 14:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 16:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 14:39:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-22 10:14:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"069c20a5b7d1279261604819c38ce5c35944debd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-19 16:28:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0725cc496fb3bb3f1a97628256c1fcb778367cc8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 16:30:03.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"990bbe3b5e543640a218c16deeebc3bc0f4824d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 16:41:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883dcc7932af49ac84b6950438e05ac9 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e53f75eeb85843c7badbc5da1c4a3f49 : FAILURE in 12m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/61c9e5eaaad24800bd0c6ce0f8b614cd : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"39efb856a74bdf1c5ea1025a503410f7fc8d7f3e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-20 12:22:47.000000000","message":"Patch Set 1:\n\nrecheck\n\nthe depends-on merged should now be OK","accounts_in_message":[],"_revision_number":1},{"id":"69a684c4c5152c301ec35a0da9208e00f657c9d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 12:35:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c1d623ac7e4d02a04ef4ec2660331a : SUCCESS in 9m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f40d0580e2d244e6960adc089492f3c3 : FAILURE in 11m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2b38d3d4994d4c3a8fa2aef7891e5a5d : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2d7f50061af416e1c2f4768153823f59d2adbcf6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-20 12:43:04.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nadding links for reviewers","accounts_in_message":[],"_revision_number":1},{"id":"49f28ae9ded29ca40c0228aa2de9a6a75a365e2a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-20 12:45:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f038ee0d4c2205953b14e7134c35543eab90a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-20 12:46:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1c01feb7513e35e613c0a54ebda9c656710d7e31","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 12:47:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"513ac1c486682f5cc15217d9c69f7ce5f271bbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 13:04:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491023589ebf41c6a679398aa8e491db : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60c83e2fa21f4c30b7c15e863f8c148b : SUCCESS in 12m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5e347bb9b99942e0a2bb198def628a09 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"1ed975bed03564e2cd2f13f5a4738e1a20d7cb5a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-21 14:39:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nI have one question inline, giving -1 to call attention to it.","accounts_in_message":[],"_revision_number":2},{"id":"664e96c90d83e089fd081ded09eeeb65f60fb6a6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-21 14:44:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bff40a3d339a7805ef7f036b3712ff2729f75211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-21 14:44:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"de55c52211cc1a2b5dce834122a02fb11b5a423f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 14:45:52.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"ceb6c31fb6e0bbb5ac6c9f607224f019bf7e3448","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-21 14:56:14.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the update! LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"d6075a8613bf9ba253d4e4a78404bd8c8eb936f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:00:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc98b4ec6a614b56aae8eaa745dc196c : SUCCESS in 9m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab90bd9af2fd4e149b72e10defcec7b4 : SUCCESS in 11m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7d1107b813b4434c8adbd13b1c0d2652 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cde9c36e9a80425233a3bb46af0c71a142c579a4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-07-22 10:14:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c9877c39668bea644b49c6b5df22a16d9eb01299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 10:15:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e7939fac98a01bd60498a5a2b7770680a5a6866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 10:33:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87b6abd57e4463caecd5cb8b7786797 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3de110139c24e12a986a0868142604f : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3313f293b11f4131ab69de6419b8a5ef : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b2b7b069dbb6963a76d4b6b546068f1486336f9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 10:33:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"720ff8122cdce4b26c1489e3ea8d01dfad343e32","revisions":{"b1666e7a26523c80d312b51622ba213e4ec2dabc":{"kind":"REWORK","_number":1,"created":"2021-07-19 16:28:34.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/53/801353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/801353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/801353/1"}}},"commit":{"parents":[{"commit":"e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5","subject":"Merge \"Release sushy 3.10 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-19 16:20:29.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-19 16:20:29.000000000","tz":180},"subject":"Makes victoria tripleo release","message":"Makes victoria tripleo release\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Mostly bumps bugfix except two that deserve\nbumping .minor.\n\nDepends-On: I2f514a1805933577173dfa57120e360cca05be10\nChange-Id: I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1666e7a26523c80d312b51622ba213e4ec2dabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1666e7a26523c80d312b51622ba213e4ec2dabc"}]},"branch":"refs/heads/master"},"1ee28c93d414dd0afbde3679b32008d97e7a6dfe":{"kind":"REWORK","_number":2,"created":"2021-07-20 12:46:35.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/53/801353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/801353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/801353/2"}}},"commit":{"parents":[{"commit":"e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5","subject":"Merge \"Release sushy 3.10 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-19 16:20:29.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-20 12:46:31.000000000","tz":0},"subject":"Makes victoria tripleo release","message":"Makes victoria tripleo release\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Mostly bumps bugfix except two that deserve\nbumping .minor.\n\nDepends-On: I2f514a1805933577173dfa57120e360cca05be10\nChange-Id: I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ee28c93d414dd0afbde3679b32008d97e7a6dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ee28c93d414dd0afbde3679b32008d97e7a6dfe"}]},"branch":"refs/heads/master"},"720ff8122cdce4b26c1489e3ea8d01dfad343e32":{"kind":"REWORK","_number":3,"created":"2021-07-21 14:44:50.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/53/801353/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/801353/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/801353/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/801353/3"}}},"commit":{"parents":[{"commit":"e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5","subject":"Merge \"Release sushy 3.10 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1b3fa10962cefad3220ae41e1c81a0ae0fd0fd5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-19 16:20:29.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-21 14:44:46.000000000","tz":0},"subject":"Makes victoria tripleo release","message":"Makes victoria tripleo release\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Mostly bumps bugfix except two that deserve\nbumping .minor.\n\nDepends-On: I2f514a1805933577173dfa57120e360cca05be10\nChange-Id: I9495cb36a9ac8f8dcb225c2c5839ab7b53d341de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/720ff8122cdce4b26c1489e3ea8d01dfad343e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/720ff8122cdce4b26c1489e3ea8d01dfad343e32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
