)]}'
{"id":"openstack%2Freleases~802057","triplet_id":"openstack%2Freleases~master~I2708a7fb4d38f9f2528010fe9294ac66fa8b12bf","project":"openstack/releases","branch":"master","topic":"ussuri-stable","hashtags":[],"change_id":"I2708a7fb4d38f9f2528010fe9294ac66fa8b12bf","subject":"Release glance for stable/ussuri","status":"MERGED","created":"2021-07-23 16:37:49.000000000","updated":"2021-08-02 10:04:09.000000000","submitted":"2021-08-02 10:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802057-ussuri-stable","meta_rev_id":"83aca72f1f6f8f63f54615246a7cfe13abdc8189","_number":802057,"virtual_id_number":802057,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 10:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-08-02 09:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-08-02 09:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-08-02 09:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-02 09:10:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 16:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 16:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 17:20:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-07-23 17:20:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-02 09:20:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"e55352e167f6d32e280d5aec62893a597b28eb6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-23 16:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc97ef5cfa9d363db650b2425ed1e009f0158f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 16:54:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35bca9855714fdb890b8b58e8b8a882 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbb3b5faf91545c09cf5aab4ee833fc5 : SUCCESS in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b9e7d9dd11f645d89882e618be040dcf : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8a53f7a517929452f3d68aa6bd72ff27afb90cb5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-24 16:25:29.000000000","message":"Patch Set 1: Code-Review-1\n\nI think it will be better to get below patch merged before releasing stable/ussuri glance.\n\nhttps://review.opendev.org/c/openstack/glance/+/785552","accounts_in_message":[],"_revision_number":1},{"id":"021a8778386df79a8a8931a93d3f0ee4b38a9d34","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-25 15:26:09.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e I think it will be better to get below patch merged before releasing stable/ussuri glance.\n\u003e \n\u003e https://review.opendev.org/c/openstack/glance/+/785552\n\nThanks Abhishek! Feel free to update the patch whenever the linked patch has merged.","accounts_in_message":[],"_revision_number":1},{"id":"ff2d4cf132ed5e7a7d2d1531f257a0340c33612a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 09:09:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3e08674a37236cbe20be2cd34bdffa1122778543","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 09:10:56.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"2d21fa2ed267f50473db63fbfd7f15eeb094d8ba","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-02 09:20:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"31e569b28c470caad03e8dcc726f16132b4df024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 09:22:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effac9e8b09d47c1957d05787c5b314c : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94a77e97af7440c391224fcdfebf79b2 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ac13bddcaa674da8bc0a555f0494a147 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"bcdf32737ed338200a8e63bc7aeb84e5a8ba9472","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-02 09:51:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for the update, LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"422c84f05c1aa9b7eb274768e5954caeef4577de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 09:52:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13bb716fc12ff6169f104eba65412d0790a40270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:04:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d59d991a5f4032a8ac4fcceef8d1ad : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c70095f3ee4a63afba54deb65fa048 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3a877937cb645db84dba87ffb72a09e : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"83aca72f1f6f8f63f54615246a7cfe13abdc8189","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:04:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8493fc103ec7434a9e7d8bcc59980b65647e7ed","revisions":{"82fa23304f0b625393fb4d34e7989796b5febd03":{"kind":"REWORK","_number":1,"created":"2021-07-23 16:37:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/57/802057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/802057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/802057/1"}}},"commit":{"parents":[{"commit":"3de9ff8b45aa94462127359d87ab1d0ad1542995","subject":"Merge \"Tag stable/stein tripleo repos as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3de9ff8b45aa94462127359d87ab1d0ad1542995"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 16:37:44.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 16:37:44.000000000","tz":120},"subject":"Release glance for stable/ussuri","message":"Release glance for stable/ussuri\n\nThis release picks up new commits to glance since\nthe last release from stable/ussuri.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 20.0.1..cc22468d\ncc22468d Drop lower-constraints jobs\n977ce4d7 Fix a failure to parse json file\n3880298a Add a release note about import locking\n5b681da1 Cleanup import status information after busting a lock\n03a11dbb Add ImageLock to base flow checks\n6c8f1e81 Functional test enhancement for lock busting\n7ba0762d Add functional test for task status updating\nbb573185 Handle atomic image properties separately\nbeece27e Move SynchronousAPIBase to a generalized location\n173fdb19 Add FakeData generator test utility\n28773650 Fix metadefs for compute-watchdog\n055e5e79 Implement time-limited import locking\n5998933a Add context.elevated() helper for getting admin privileges\nebeb31e6 Poll for final state on test_copy_image_revert_lifecycle()\n541c0fd6 Fix non-deterministic copy_image_revert_lifecycle test\n2d5e8439 Fix import failure status reporting when all_stores_must_succeed\u003dTrue\n931d39ec Functional reproducer for bug 1891352\n3da696e0 Update task message during import\n586ca78a Heartbeat the actual work of the task\n825a0ec7 Add image_delete_property_atomic() helper\n56d01f25 Add image_set_property_atomic() helper\n663e027e Flesh out FakeImage for extra_properties\nec46cbca Add tests for _ImportToStore.execute()\nea34be42 Add testing for _CompleteTask in api_image_import\n0895d191 Make test_copy_image_revert_lifecycle handle 409 on import retry\neb08fbae Make import task capable of running as admin on behalf of user\n168fdb7f Add a functional test for non-owned image copying\n3bf1af9b Refactor common auth token code in images test\nbc826947 Add a test to replicate the owner-required behavior of copy-image\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2708a7fb4d38f9f2528010fe9294ac66fa8b12bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/82fa23304f0b625393fb4d34e7989796b5febd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/82fa23304f0b625393fb4d34e7989796b5febd03"}]},"branch":"refs/heads/master"},"b8493fc103ec7434a9e7d8bcc59980b65647e7ed":{"kind":"REWORK","_number":2,"created":"2021-08-02 09:09:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/802057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/802057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/802057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/802057/2"}}},"commit":{"parents":[{"commit":"3de9ff8b45aa94462127359d87ab1d0ad1542995","subject":"Merge \"Tag stable/stein tripleo repos as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3de9ff8b45aa94462127359d87ab1d0ad1542995"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 16:37:44.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-02 09:09:29.000000000","tz":0},"subject":"Release glance for stable/ussuri","message":"Release glance for stable/ussuri\n\nThis release picks up new commits to glance since\nthe last release from stable/ussuri.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 20.0.1..cc22468d\ncc22468d Drop lower-constraints jobs\n977ce4d7 Fix a failure to parse json file\n3880298a Add a release note about import locking\n5b681da1 Cleanup import status information after busting a lock\n03a11dbb Add ImageLock to base flow checks\n6c8f1e81 Functional test enhancement for lock busting\n7ba0762d Add functional test for task status updating\nbb573185 Handle atomic image properties separately\nbeece27e Move SynchronousAPIBase to a generalized location\n173fdb19 Add FakeData generator test utility\n28773650 Fix metadefs for compute-watchdog\n055e5e79 Implement time-limited import locking\n5998933a Add context.elevated() helper for getting admin privileges\nebeb31e6 Poll for final state on test_copy_image_revert_lifecycle()\n541c0fd6 Fix non-deterministic copy_image_revert_lifecycle test\n2d5e8439 Fix import failure status reporting when all_stores_must_succeed\u003dTrue\n931d39ec Functional reproducer for bug 1891352\n3da696e0 Update task message during import\n586ca78a Heartbeat the actual work of the task\n825a0ec7 Add image_delete_property_atomic() helper\n56d01f25 Add image_set_property_atomic() helper\n663e027e Flesh out FakeImage for extra_properties\nec46cbca Add tests for _ImportToStore.execute()\nea34be42 Add testing for _CompleteTask in api_image_import\n0895d191 Make test_copy_image_revert_lifecycle handle 409 on import retry\neb08fbae Make import task capable of running as admin on behalf of user\n168fdb7f Add a functional test for non-owned image copying\n3bf1af9b Refactor common auth token code in images test\nbc826947 Add a test to replicate the owner-required behavior of copy-image\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2708a7fb4d38f9f2528010fe9294ac66fa8b12bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8493fc103ec7434a9e7d8bcc59980b65647e7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8493fc103ec7434a9e7d8bcc59980b65647e7ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
