)]}'
{"id":"openstack%2Freleases~802109","triplet_id":"openstack%2Freleases~master~I8c97acf6a72417363a6192a0e36728b9592257b6","project":"openstack/releases","branch":"master","topic":"add-pypi-template","hashtags":[],"change_id":"I8c97acf6a72417363a6192a0e36728b9592257b6","subject":"Release panko for stable/ussuri","status":"MERGED","created":"2021-07-23 20:12:11.000000000","updated":"2021-08-04 12:01:09.000000000","submitted":"2021-08-04 12:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802109-add-pypi-template","meta_rev_id":"a6537360bd0d8aef83aab35484b237cd28fdc6ba","_number":802109,"virtual_id_number":802109,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-04 12:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-08-04 11:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-08-04 12:01:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-08-04 09:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-08-04 11:49:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-04 12:01:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 20:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 20:20:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-07-23 20:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-04 08:37:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-08-04 09:23:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-08-04 09:30:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-04 11:49:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"db1cc58dbbd198d1abc80f09fc883f55ca03e0cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-23 20:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a446eb9dae99a01f9aececc9516e2f65c65fb360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:24:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75b6984d2ff4ed0bb675b108e72742a : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54aeeeeae50c4f2f99636d86f8707fbe : FAILURE in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cc4b930cb5fb438684cdb2454446699c : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6e4ef022de6073483d59eda6769c106ebf53b918","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-27 08:05:49.000000000","message":"Patch Set 1: Code-Review+1\n\nThank you, looks good in general.\n\nThis seem to fail because of the release-to-pypi job is missing?","accounts_in_message":[],"_revision_number":1},{"id":"026aab74fe42ceb3a11e1f190946c22a7a9ce426","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:06:53.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cf7f307e508b5c51424bcac221725427781e3625","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-27 13:04:17.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e Thank you, looks good in general.\n\u003e \n\u003e This seem to fail because of the release-to-pypi job is missing?\n\nYes it is failing because of that [1] I\u0027ll double check with the team about how we could make it pass.\n\n[1] https://zuul.opendev.org/t/openstack/build/54aeeeeae50c4f2f99636d86f8707fbe/log/tox/validate-request-results.log#135","accounts_in_message":[],"_revision_number":1},{"id":"dcf43963036c517691b1123c9eba97ffde20fdca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-04 08:37:32.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e Thank you, looks good in general.\n\u003e \u003e \n\u003e \u003e This seem to fail because of the release-to-pypi job is missing?\n\u003e \n\u003e Yes it is failing because of that [1] I\u0027ll double check with the team about how we could make it pass.\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/54aeeeeae50c4f2f99636d86f8707fbe/log/tox/validate-request-results.log#135\n\nWeird...\n\n`publish-to-pypi-stable-only` is well defined in the project-config for panko.\n\nI compared the configurations of paunch and panko, those projects are in the same situation, they are no longer maintained but still with open stable branches, and I don\u0027t saw significant differences between [1][2] them that could explain why on one side (panko) we fail with an undefined `publish-to-pypi` and why on the other side the same config works (publish-to-pypi-stable-only).\n\nThe only difference that I found is that  paunch declare `release-type: python-pypi` on Ussuri [3].\n\nOn the gouvernance side, I found few differences between these projects. Indeed, paunch set `release-management: deprecated` [4] and that\u0027s not the case with panko [5].\n\nI would suggest to start by trying to add `release-type: python-pypi` into panko\u0027s ussuri deliverable file [6]. If that doesn\u0027t help us, then I would suggest to set ` release-management: deprecated` on the panko config into the governance repo [5].\n\nLet\u0027s see what will happen with that.\n\n[1] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L4354-L4366\n[2] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L4374-L4388\n[3] https://opendev.org/openstack/releases/src/branch/master/deliverables/ussuri/paunch.yaml#L4\n[4] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L3089\n[5] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L3010-L3015\n[6] https://opendev.org/openstack/releases/src/branch/master/deliverables/ussuri/panko.yaml","accounts_in_message":[],"_revision_number":1},{"id":"42ea31a2f228148760196d9fc92451a9ca62e37e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-04 09:23:29.000000000","message":"Patch Set 1:\n\nI suspect this is the first time we do a panko/paunch release since d7304b6416 merged on project-config:\n\nhttps://opendev.org/openstack/project-config/commit/d7304b6416b7596fc3221f5989758db79642f8a9\n\nWe probably need to add publish-to-pypi-stable-only to https://opendev.org/openstack/releases/src/branch/master/openstack_releases/project_config.py#L138","accounts_in_message":[],"_revision_number":1},{"id":"2a44dbe4e247ec5505ab07b68c8b642e37bac69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-04 09:30:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e39c51593b9af429567ba294374172be158e9279","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-04 09:32:43.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I suspect this is the first time we do a panko/paunch release since d7304b6416 merged on project-config:\n\u003e \n\u003e https://opendev.org/openstack/project-config/commit/d7304b6416b7596fc3221f5989758db79642f8a9\n\u003e \n\u003e We probably need to add publish-to-pypi-stable-only to https://opendev.org/openstack/releases/src/branch/master/openstack_releases/project_config.py#L138\n\nGood catch!\n\nI added it to our config https://review.opendev.org/c/openstack/releases/+/803452 and this patch is now rebased on the top of that.","accounts_in_message":[],"_revision_number":2},{"id":"74763c235e563382a8151547eceb638321dc09a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 09:39:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5301054bc4eb401980887911ddbb2f98 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a4278117957432196c2b436bc3ebc1c : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/545e9ade9716489a9869b30430443d57 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8e66d355a1992f87158694483a8f28345e4370c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-04 09:42:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9c544c6b8b86ac542cf16288e833cbe78a40f970","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-04 11:49:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6e0b9c521137ed396f40380ae96d295bbf87579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 11:50:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"983c24c0764d7c6e63888eedeafb3377e6eeee31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 12:01:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d150eacd844f2cadfab5600e534b9f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a051b399a3ef451885a9ec42fffef8b6 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6eabb6fc2b2245128bb59d00a62d1811 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e04a7f6457d803c5d33f89d2b2392e8ae7281d30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 12:01:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f9ccae01d129aab1b7fa0c833f86f5cb1a5d06f","revisions":{"0a2a71820a16f0715e5c4f0144a55444bc36f648":{"kind":"REWORK","_number":1,"created":"2021-07-23 20:12:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/09/802109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/802109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/802109/1"}}},"commit":{"parents":[{"commit":"3de9ff8b45aa94462127359d87ab1d0ad1542995","subject":"Merge \"Tag stable/stein tripleo repos as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3de9ff8b45aa94462127359d87ab1d0ad1542995"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 20:12:05.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 20:12:05.000000000","tz":120},"subject":"Release panko for stable/ussuri","message":"Release panko for stable/ussuri\n\nThis release picks up new commits to panko since\nthe last release from stable/ussuri.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 8.0.0..0a8d5b4f\n0a8d5b4f Add max_count argument for clear_expired_data call\n4e8597a1 Update TOX_CONSTRAINTS_FILE for stable/ussuri\nc47b5791 Update .gitreview for stable/ussuri\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8c97acf6a72417363a6192a0e36728b9592257b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a2a71820a16f0715e5c4f0144a55444bc36f648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a2a71820a16f0715e5c4f0144a55444bc36f648"}]},"branch":"refs/heads/master"},"7f9ccae01d129aab1b7fa0c833f86f5cb1a5d06f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-04 09:30:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/09/802109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/802109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/802109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/802109/2"}}},"commit":{"parents":[{"commit":"fa46c1fec4874b373664e3350f65048c22166367","subject":"Adding support of publish-to-pypi-stable-only template for deprecated repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa46c1fec4874b373664e3350f65048c22166367"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 20:12:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-04 09:29:42.000000000","tz":120},"subject":"Release panko for stable/ussuri","message":"Release panko for stable/ussuri\n\nThis release picks up new commits to panko since\nthe last release from stable/ussuri.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 8.0.0..0a8d5b4f\n0a8d5b4f Add max_count argument for clear_expired_data call\n4e8597a1 Update TOX_CONSTRAINTS_FILE for stable/ussuri\nc47b5791 Update .gitreview for stable/ussuri\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8c97acf6a72417363a6192a0e36728b9592257b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7f9ccae01d129aab1b7fa0c833f86f5cb1a5d06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7f9ccae01d129aab1b7fa0c833f86f5cb1a5d06f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
