)]}'
{"id":"openstack%2Freleases~802123","triplet_id":"openstack%2Freleases~master~I14dc477e0e5b77c1ec7b2b4031407d7252d85891","project":"openstack/releases","branch":"master","topic":"ussuri-stable","hashtags":[],"change_id":"I14dc477e0e5b77c1ec7b2b4031407d7252d85891","subject":"Release swift for stable/ussuri","status":"MERGED","created":"2021-07-23 20:50:26.000000000","updated":"2021-09-01 08:28:18.000000000","submitted":"2021-09-01 08:28:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802123-ussuri-stable","meta_rev_id":"6db45cf69ad79482488c8075f1fc5323fb4ae665","_number":802123,"virtual_id_number":802123,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 08:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-08-31 20:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-01 08:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-31 18:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-01 08:02:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-31 18:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 20:51:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 21:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 21:20:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-08-30 07:23:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-09-01 08:02:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"07056156817b27aa5ad8b337b31661a812259391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-23 20:50:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"045b9a0b79545062de07b53f613d817b7087d9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 21:10:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8186bf35f49e4582bfe7aea3a1a09ded : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de4390eba28f48e6a7cb331154b7d28e : SUCCESS in 10m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4fa2e750e101409cbc831c3ad7e1d893 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"35d729095e302dd2b27ac5e7a13ff556d6458c06","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 07:23:39.000000000","message":"Patch Set 1:\n\nHello Swift team,\n\nPlease can you have a look?","accounts_in_message":[],"_revision_number":1},{"id":"1a70abb4b9593f75f2cc4d8b038663eec66dc508","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-31 18:43:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ba162a61d8ea876fc02a23e75c7f51c32c8b565","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:44:29.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2d73aefbe8ace6a87f434933d077c19d44825996","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-31 20:40:45.000000000","message":"Patch Set 1: Code-Review+2\n\nApproved by PTL (thanks!)","accounts_in_message":[],"_revision_number":1},{"id":"9d32c229a1292148d46958ed9ff0cb8817179f62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 08:02:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a19585aca349f72c6c92a9ff29390b24314df382","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 08:04:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c7595f0f2208e433206843a6852e2a47ba49d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 08:28:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b5781d4999440da142d04faa20e3c5 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1dda5d1efb74885b59212b25788bb1a : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2675ef5f999148bbb20d43fec5de1376 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6db45cf69ad79482488c8075f1fc5323fb4ae665","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 08:28:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af49ea98d95411c81863f90ffcc4570c944e4794","revisions":{"af49ea98d95411c81863f90ffcc4570c944e4794":{"kind":"REWORK","_number":1,"created":"2021-07-23 20:50:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/802123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/802123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/802123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/802123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/802123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/802123/1"}}},"commit":{"parents":[{"commit":"3de9ff8b45aa94462127359d87ab1d0ad1542995","subject":"Merge \"Tag stable/stein tripleo repos as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3de9ff8b45aa94462127359d87ab1d0ad1542995"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 20:50:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-07-23 20:50:20.000000000","tz":120},"subject":"Release swift for stable/ussuri","message":"Release swift for stable/ussuri\n\nThis release picks up new commits to swift since\nthe last release from stable/ussuri.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 2.25.1..4b0d6a87e\n29eefcd0d Dep\u0027s should be restricted by upper-constraints\n18d8c15f9 symlink: Ensure headers are returned as a list on py3\n4a38f62b2 tempurl: Fix PUT upload to temp url on py3\ne69b3c60f Avoid loops when gathering container listings from shards\n18cdf9b56 Populate shrinking shards with shard ranges learnt from root\n5791cec31 Use swiftclient Connections in sharding probe tests\n959a98d7b Refactor audit shard tests\n760528cb7 container-sharding: Stable overlap order\n94346b134 Address a sharder/replicator race\neef68d69c Identify deleted shards as shards\nf4061c043 swift-container-info: Show shard ranges summary\n8588296dc Don\u0027t auto-create shard containers\n0da122383 Latch shard-stat reporting\nd1b997889 py27: Suppress UnicodeWarnings in ShardRange setters\n1b3b77696 sharding: Add probe test that exercises swift-manage-shard-ranges\n9491d73a6 Give probe tests a second chance to pass\n69c93dcb1 Bump up probe test timeout\nebc023f53 Bump up timeouts for DSVM and probe test jobs\n067328c8d Give unit tests a second chance to pass\n2e487b42d s3-tests: Use my s3compat fork instead\n9820aad4a s3-tests: Only try to copy outputs when there *are* outputs\na357fc66a sharding: Don\u0027t inject shard ranges when user says quit\n69a95a52a Prevent upgrading to pip 21+\nb6d320b1d Memcached client TLS support\n41bec521f [stable-only] Cap bandit to 1.6.2\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I14dc477e0e5b77c1ec7b2b4031407d7252d85891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af49ea98d95411c81863f90ffcc4570c944e4794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af49ea98d95411c81863f90ffcc4570c944e4794"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
