)]}'
{"id":"openstack%2Freleases~803219","triplet_id":"openstack%2Freleases~master~I45e0555af221bfe58f9f014e4d30e813610dddf9","project":"openstack/releases","branch":"master","topic":"not-yet-released","hashtags":[],"change_id":"I45e0555af221bfe58f9f014e4d30e813610dddf9","subject":"[neutron] Propose to switch the release model","status":"ABANDONED","created":"2021-08-02 14:35:45.000000000","updated":"2021-08-03 15:43:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e39d02d933ae0228d0a07e6408eb165bf9cacb61","_number":803219,"virtual_id_number":803219,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-02 14:43:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":-1,"date":"2021-08-03 09:18:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":-1,"date":"2021-08-03 09:00:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 14:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-02 14:43:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-02 15:20:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-08-02 15:20:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-03 07:49:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-08-03 09:18:46.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"ccb68113979616c9f6cf5c59e8326adab75e1607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-02 14:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8958efc8f73cfa9cac45c9dff193eccb41a2889a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 14:43:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a7c671cec34acf8dea59c999a5ba3c : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5856c3920204fec81485b2b0f8b0067 : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/07b895de55dc4273a05d53e647b7ac87 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a6f845b14921c22d082b88ad2a81fb12ab5f96be","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-03 09:00:43.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m not sure if that will be good change really as we may need to release this more often in the future but I\u0027m open to others\u0027 opinions 😊\nFor now I proposed new release in https://review.opendev.org/c/openstack/releases/+/803321","accounts_in_message":[],"_revision_number":1},{"id":"1d6fb9b6e792b07b2fd6bc27cf02701924a4541d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-03 09:18:46.000000000","message":"Patch Set 1: Code-Review-1\n\nHi Hervé,\n\nI am voting -1 because I think that for now we want to keep the ability of releasing more often during the development cycle. That said, there\u0027s some reorganization going on right for the ppl that is maintaining this project as the main maintainer is not longer around, so I expect a few changes and perhaps later we may opt for the release-with-rc model but, for now I think that intermediary release model is a good thing to keep.","accounts_in_message":[],"_revision_number":1},{"id":"32ef7e90d8e053da4b6ffe760822512d51389332","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-03 15:43:18.000000000","message":"Patch Set 1:\n\nOk thanks guys no problem. This patch aimed to highlight this point to open the discussion.\nI\u0027ll abandon this patch now that you expressed your thoughts.\n\nThank you","accounts_in_message":[],"_revision_number":1},{"id":"e39d02d933ae0228d0a07e6408eb165bf9cacb61","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-03 15:43:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07371fe664392fad89b32e7ca9fce164f79a2ad7","revisions":{"07371fe664392fad89b32e7ca9fce164f79a2ad7":{"kind":"REWORK","_number":1,"created":"2021-08-02 14:35:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/803219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/803219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/803219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/803219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/803219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/803219/1"}}},"commit":{"parents":[{"commit":"a964fe144554d9857ece489e0db00421d1837f29","subject":"[heat] Propose to switch the release model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a964fe144554d9857ece489e0db00421d1837f29"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-02 14:31:08.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-02 14:31:08.000000000","tz":120},"subject":"[neutron] Propose to switch the release model","message":"[neutron] Propose to switch the release model\n\nIntermediary-released deliverables that did release only once during\nthe last cycle, and have not done a release yet are good candidates\nto switch to the cycle-with-rc model, which is much more suitable for\ndeliverables that are only released once per cycle.\n\nThese changes are not mandatory and this patch is a conveniences to\nhighlight this point and open the discussion.\n\nPTLs and release liaisons please decide to:\n- immediately release an intermediary release (and -1 the proposed change)\n- confirm the release model change (+1 the proposed change)\n- stay uncertain for this cycle of how many releases will be made, but\n  acknowledge that they need to do a release before RC1 (-1 the proposed\n  change)\n\nChange-Id: I45e0555af221bfe58f9f014e4d30e813610dddf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07371fe664392fad89b32e7ca9fce164f79a2ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07371fe664392fad89b32e7ca9fce164f79a2ad7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
