)]}'
{"id":"openstack%2Freleases~803220","triplet_id":"openstack%2Freleases~master~I4513b9e5b03f0a7da9277ff9a2e04c9a7ba75080","project":"openstack/releases","branch":"master","topic":"not-yet-released","hashtags":[],"change_id":"I4513b9e5b03f0a7da9277ff9a2e04c9a7ba75080","subject":"[vitrage] Propose to switch the release model","status":"ABANDONED","created":"2021-08-02 14:35:45.000000000","updated":"2021-08-03 07:51:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc1c3f75db2e68fc9ac4d94e8262ac4581933e7b","_number":803220,"virtual_id_number":803220,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-02 14:47:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-08-02 15:47:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 14:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-02 14:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-02 15:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"6aad91ca913cd049fad6a277dc9b87a01a95cc3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-02 14:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a279da07bb043414b8a2008ab51be73ffeee601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 14:47:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5363fd42f884d5bb2f75d48dca8c878 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/038aecad996b493a81a53a898151483f : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a1916522ae0d4841b70254e6b2d4a124 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"20f8a66d4ce0bd561c19bf6505117d44f861f25e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-08-02 15:47:16.000000000","message":"Patch Set 1: Code-Review-1\n\nreleased an intermediary release\nhttps://review.opendev.org/c/openstack/releases/+/803223/","accounts_in_message":[],"_revision_number":1},{"id":"bc1c3f75db2e68fc9ac4d94e8262ac4581933e7b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-03 07:51:34.000000000","message":"Abandoned\n\nThank you, I abandon this patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"112955c5460c979cb6b778c707bc3e8883de7174","revisions":{"112955c5460c979cb6b778c707bc3e8883de7174":{"kind":"REWORK","_number":1,"created":"2021-08-02 14:35:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/20/803220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/803220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/803220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/803220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/803220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/803220/1"}}},"commit":{"parents":[{"commit":"07371fe664392fad89b32e7ca9fce164f79a2ad7","subject":"[neutron] Propose to switch the release model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07371fe664392fad89b32e7ca9fce164f79a2ad7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-02 14:34:19.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-02 14:34:19.000000000","tz":120},"subject":"[vitrage] Propose to switch the release model","message":"[vitrage] Propose to switch the release model\n\nIntermediary-released deliverables that did release only once during\nthe last cycle, and have not done a release yet are good candidates\nto switch to the cycle-with-rc model, which is much more suitable for\ndeliverables that are only released once per cycle.\n\nThese changes are not mandatory and this patch is a conveniences to\nhighlight this point and open the discussion.\n\nPTLs and release liaisons please decide to:\n- immediately release an intermediary release (and -1 the proposed change)\n- confirm the release model change (+1 the proposed change)\n- stay uncertain for this cycle of how many releases will be made, but\n  acknowledge that they need to do a release before RC1 (-1 the proposed\n  change)\n\nChange-Id: I4513b9e5b03f0a7da9277ff9a2e04c9a7ba75080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/112955c5460c979cb6b778c707bc3e8883de7174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/112955c5460c979cb6b778c707bc3e8883de7174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
