)]}'
{"id":"openstack%2Freleases~803885","triplet_id":"openstack%2Freleases~master~I92f709b4241f8ae3c1165627db441756185b683e","project":"openstack/releases","branch":"master","topic":"puppet-openstack-xena-m2","hashtags":[],"change_id":"I92f709b4241f8ae3c1165627db441756185b683e","subject":"Puppet OpenStack Xena Milestone 2","status":"MERGED","created":"2021-08-09 08:50:30.000000000","updated":"2021-08-16 08:20:45.000000000","submitted":"2021-08-16 08:20:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"803885-puppet-openstack-xena-m2","meta_rev_id":"75bbe6afb410d88c968bd99d4335ebab86cf4fb2","_number":803885,"virtual_id_number":803885,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 08:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-08-16 08:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-08-10 06:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-08-16 08:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-09 08:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-09 08:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 13:25:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-08-10 06:17:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-16 08:06:59.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"a50195800686ff5ff0e1318c1b09f71b971585f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-08-09 08:50:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ada5712e591402196e3a5096a35f5038c3500e90","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 08:51:53.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5da3ace6513fdb7e809e5e5b171d02d5e9613b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 08:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3782ac9c9dd04c19b1229b876ddb9c79 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6e04c643490418d9cb50fe6576d4cb9 : SUCCESS in 7m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e311839edfd48c58a270c561fbb986e : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"806fcdf2bd9397124abd2178fc22e32bbcb7d851","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-09 13:25:51.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHello,\n\nI seen that almost all your projects removed fedora\u0027s support.\nI suppose that we need a major version to reflect this change.\n\nAlso I noticed some other removal that will require a major version, see my inline comments.","accounts_in_message":[],"_revision_number":1},{"id":"58cb647ee5c86c52df1b6a538246238bf9092e68","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-08-10 01:13:55.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Hello,\n\u003e \n\u003e I seen that almost all your projects removed fedora\u0027s support.\n\u003e I suppose that we need a major version to reflect this change.\n\u003e \n\u003e Also I noticed some other removal that will require a major version, see my inline comments.\n\nThis version is finished with the following patch:\nhttps://review.opendev.org/q/topic:%22prep-release-xena-m2%22+(status:open%20OR%20status:merged)\nThe new changes will be in the next version 19.2.0","accounts_in_message":[],"_revision_number":1},{"id":"342dc0a46ee4bdf8a41050379b31efbbbfbe3d55","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 06:17:00.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e (2 comments)\n\u003e \u003e \n\u003e \u003e Hello,\n\u003e \u003e \n\u003e \u003e I seen that almost all your projects removed fedora\u0027s support.\n\u003e \u003e I suppose that we need a major version to reflect this change.\n\u003e \u003e \n\u003e \u003e Also I noticed some other removal that will require a major version, see my inline comments.\n\u003e \n\u003e This version is finished with the following patch:\n\u003e https://review.opendev.org/q/topic:%22prep-release-xena-m2%22+(status:open%20OR%20status:merged)\n\u003e The new changes will be in the next version 19.2.0\n\nYes I know that we finished with the patch series above. Do you mean that we don\u0027t need a major bump and that the given versions are correct?\n\nAnyway, I won\u0027t ask you to resubmit 40 patches to re-update the versions. The fedora changes looks minor changes that nobody use.","accounts_in_message":[],"_revision_number":1},{"id":"9b3e06ce1ef3daca231a3e721b271ceb2c564ea6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-08-16 01:18:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cff30916cf9ef31172b7a1facef914246a8f2fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 01:30:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39dcaceba6124b12b4d84f21f3dae957 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d585f2674d0c486589cc6991a7286883 : SUCCESS in 9m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3835556caf5945c580c1821afe2e2d29 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"74e3b39a93bacf6bc03901a17bad0b3fb03e8bd2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-16 08:06:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cb56d5d689bb98410feb42ab75289c7f9cb39d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 08:07:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"959b61870b63a62f97156b06c9303486bfac3dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 08:20:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8bd5c0887864a6c9a9a465d0b47ccd0 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823eb5972df545ee9ce249b63ec56d0c : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21c734057a3848a890873be4e4fda4fc : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"75bbe6afb410d88c968bd99d4335ebab86cf4fb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 08:20:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0de6d778a73eab487ca42b3ace68644cf5c323c","revisions":{"c0de6d778a73eab487ca42b3ace68644cf5c323c":{"kind":"REWORK","_number":1,"created":"2021-08-09 08:50:30.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/85/803885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/803885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/803885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/803885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/803885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/803885/1"}}},"commit":{"parents":[{"commit":"8d125adba23a1741a948fc206cc5188581b24ea1","subject":"Merge \"[designate-tempest-plugin] Tag train-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d125adba23a1741a948fc206cc5188581b24ea1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-08-09 08:46:40.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-08-09 08:46:40.000000000","tz":480},"subject":"Puppet OpenStack Xena Milestone 2","message":"Puppet OpenStack Xena Milestone 2\n\nChange-Id: I92f709b4241f8ae3c1165627db441756185b683e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0de6d778a73eab487ca42b3ace68644cf5c323c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0de6d778a73eab487ca42b3ace68644cf5c323c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
