)]}'
{"id":"openstack%2Freleases~804105","triplet_id":"openstack%2Freleases~master~I3a2b5e7296fefa26c100f7726ce364e7179e62e2","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I3a2b5e7296fefa26c100f7726ce364e7179e62e2","subject":"Make wallaby release for all tripleo repos","status":"MERGED","created":"2021-08-10 15:50:35.000000000","updated":"2021-08-16 07:51:47.000000000","submitted":"2021-08-16 07:51:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"804105","meta_rev_id":"cfe3dcb3a4fb8b04624e6568ebb7417f2ed3fc2e","_number":804105,"virtual_id_number":804105,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 07:51:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-08-13 12:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-08-16 07:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-13 11:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-08-16 07:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-13 11:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 15:52:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 11:35:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-08-13 11:55:36.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-08-13 12:23:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-08-16 07:38:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b9c11f6b0d3faaec73a59683b39e552a3e9bd7c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-10 15:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58c0ca6e378c44e1903b452dc361574ea4f219e2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-10 15:51:47.000000000","message":"Patch Set 1: Workflow-1\n\nneeds depends-on first\n\nand likely update tripleo-heat-templates and tripleo-ansible for bumping minor instead of .patch","accounts_in_message":[],"_revision_number":1},{"id":"32e503ccca495fc264f5b6c1fb21c838e499cf65","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:52:04.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"7b79c6f9de8265d8c66a75601531ca1521b260d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:17:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e014bf317a4d42a1080331432fb7cc : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a532446311e149e7b6e44e5afdfa13e0 : FAILURE in 16m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c33332c9be7140c3af860354f5b000e5 : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fd386baa64e23bec8167df7cdf7235c3c7d601eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 11:35:13.000000000","message":"Patch Set 1:\n\nrecheck\n\nDepends-on merged","accounts_in_message":[],"_revision_number":1},{"id":"bfa75dc86fdd9ed560db3eeca861c7bdfc2a3417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 12:00:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d701045aed4e4782370b4917f8d467 : SUCCESS in 7m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a568709730143ae883f7253519e54e9 : SUCCESS in 18m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9f805de173eb492bb04416e7d7f7b64f : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5acfcf2f14182082f2a91bdf95b87b56fbe09ebb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-11 15:23:49.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e recheck\n\u003e \n\u003e Depends-on merged\n\nthanks Herve, as per my previous comment I am updating now to include .minor update for tripleo-heat-templates and tripleo-ansible i think they deserve it \n\n(.minor instead of .patch )","accounts_in_message":[],"_revision_number":1},{"id":"e634c421831c7d751112084b48aee77e29dee42e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-11 15:24:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec204a392f59555e9dd3d88e24350e51ed123ee6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 15:25:28.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"5f6b913adee75090d365e5c3c2ad94979194b8f0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-11 15:28:55.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c996983a6f677c4dc31ae28ed422c4b631ee2071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 15:34:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adedb7d468444bc88a812788ede12815 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16b4d7b6788f444ca85470f6a71d10c4 : SUCCESS in 9m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/77fbc047cfa0442c86471714a7088dc0 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"351aea8c1290405e167ac0f2b2dd2733c5ae016e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-12 14:43:37.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHello,\n\nPlease, see my inline comments.","accounts_in_message":[],"_revision_number":2},{"id":"a121013bbd24b012a637098ed747b6e1a67c77fa","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-13 11:03:19.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8fd3f32d25a3b84cd55ed9f656383eb517377eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-13 11:03:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5e157d09c65ba2df693d507ac9d537a08ef5aa2c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 11:04:26.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"bd03f9c108e8ba2c67a6df286e34aaf2ece0b1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 11:14:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6c387ee81f41fcb5f92bda203d03c2 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c531629a734940af8263299903947b : SUCCESS in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/980a9ddc2e1c44f0a2214d80f17ffb02 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ddf66357e67001509bed2a0cb206b449704a96e4","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-13 11:55:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9ec781b5206dcf042888bac324748b8b2dd7d75a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-13 12:23:26.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM:\n* version bumps look OK (Hervé\u0027s comments were addressed)\n* approved + uploaded by PTL\n* no issues seen in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"5ed9aa4f907726765cc3959b164d42b4868149f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 07:38:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f3c2a87f21c6f0c26a96a14621025e0a4a370ddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 07:38:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c585c76b62afde8526cd9ff7547203d189abb6e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 07:51:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2611e37b545345149f9cd11a62d29cdd : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170e31914c2f4b15bd7de21b023b35b0 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1c13424edc54a20a5ce72541576e87e : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":3},{"id":"cfe3dcb3a4fb8b04624e6568ebb7417f2ed3fc2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 07:51:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44c222c2f68d92a6ad5b796616daab66801b49d7","revisions":{"7b5755fb6a4b46d059d354e8624aa35a6862f9ac":{"kind":"REWORK","_number":1,"created":"2021-08-10 15:50:35.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/05/804105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/804105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/804105/1"}}},"commit":{"parents":[{"commit":"b9747e0fd6eb98dca800880f5cdb7bfc19f7a931","subject":"Merge \"Push Validation Framework release to 1.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9747e0fd6eb98dca800880f5cdb7bfc19f7a931"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-10 15:38:59.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-10 15:40:34.000000000","tz":180},"subject":"Make wallaby release for all tripleo repos","message":"Make wallaby release for all tripleo repos\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Validation jobs will fail until it merges.\n\nDepends-On: I51fdd0aa82f38c03f33175f7a238be22edba3040\nChange-Id: I3a2b5e7296fefa26c100f7726ce364e7179e62e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b5755fb6a4b46d059d354e8624aa35a6862f9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b5755fb6a4b46d059d354e8624aa35a6862f9ac"}]},"branch":"refs/heads/master"},"a342fd9284bdd39b0c7d7cf803dc6be866639423":{"kind":"REWORK","_number":2,"created":"2021-08-11 15:24:21.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/05/804105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/804105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/804105/2"}}},"commit":{"parents":[{"commit":"b9747e0fd6eb98dca800880f5cdb7bfc19f7a931","subject":"Merge \"Push Validation Framework release to 1.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9747e0fd6eb98dca800880f5cdb7bfc19f7a931"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-11 15:21:41.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-11 15:21:41.000000000","tz":180},"subject":"Make wallaby release for all tripleo repos","message":"Make wallaby release for all tripleo repos\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Validation jobs will fail until it merges.\n\nDepends-On: I51fdd0aa82f38c03f33175f7a238be22edba3040\nChange-Id: I3a2b5e7296fefa26c100f7726ce364e7179e62e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a342fd9284bdd39b0c7d7cf803dc6be866639423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a342fd9284bdd39b0c7d7cf803dc6be866639423"}]},"branch":"refs/heads/master"},"44c222c2f68d92a6ad5b796616daab66801b49d7":{"kind":"REWORK","_number":3,"created":"2021-08-13 11:03:21.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/05/804105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/804105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/804105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/804105/3"}}},"commit":{"parents":[{"commit":"b9747e0fd6eb98dca800880f5cdb7bfc19f7a931","subject":"Merge \"Push Validation Framework release to 1.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9747e0fd6eb98dca800880f5cdb7bfc19f7a931"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-11 15:21:41.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-13 11:00:54.000000000","tz":0},"subject":"Make wallaby release for all tripleo repos","message":"Make wallaby release for all tripleo repos\n\nNeeds the depends-on as we are tagging that commit here for\npuppet-tripleo. Validation jobs will fail until it merges.\n\nDepends-On: I51fdd0aa82f38c03f33175f7a238be22edba3040\nChange-Id: I3a2b5e7296fefa26c100f7726ce364e7179e62e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44c222c2f68d92a6ad5b796616daab66801b49d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44c222c2f68d92a6ad5b796616daab66801b49d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
