)]}'
{"id":"openstack%2Freleases~804412","triplet_id":"openstack%2Freleases~master~Ia1d8f84e3071bc24c6e059e4b3fcba2cc38601bc","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ia1d8f84e3071bc24c6e059e4b3fcba2cc38601bc","subject":"Make tripleo repos victoria release","status":"MERGED","created":"2021-08-12 15:58:58.000000000","updated":"2021-09-01 05:54:55.000000000","submitted":"2021-08-30 07:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"804412","meta_rev_id":"0837d63a9b4d599e81f7966e7f101eec0ab1ec03","_number":804412,"virtual_id_number":804412,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 07:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-08-27 15:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-08-26 08:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2021-08-26 13:43:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-08-27 15:10:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"CC":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 16:00:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 06:03:02.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-08-23 20:26:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-08-24 12:22:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-08-24 12:24:38.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-08-24 12:24:38.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-24 15:06:55.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"CC"},{"updated":"2021-08-24 15:46:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-08-24 15:46:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-08-27 15:10:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-08-30 06:56:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"74898c4f6bc0864cf3c7c6430a595ea82498b917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-12 15:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9537a1cc67fc9d63a55cc926ff2fa914dfbaac32","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 16:00:14.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0f5153592edebf3648cf5f690d12ed9ff5cdcd13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 16:19:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8150f806f5684888b4c56b51ab9eab91 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ec9c338dc694791aff3d58583f7b32c : FAILURE in 18m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f438c736e28249a1bca0ea05df3698e9 : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a44c696d50ff65f3541a7f99516cbb0cd243485e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-18 06:03:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77cee330f87eecb6e9199408f429990a1844102a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 20:26:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"094ad90be7b5ab2e584c3247d0d5a53dc8289329","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 20:26:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c974fd218642a3debacd5d53f9ec6e86e9e3424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 20:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f9e0e37f804bc7b475f7ccbb275874 : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/074a2ff6220b4a1a8bd561a6eb825602 : SUCCESS in 16m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1f9b2f6960ac49a5b5b58915b9091506 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c138ac8546701c22bc7c38d20ae728496662b232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-24 12:22:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebba9db034b9074841f6dc77fa1c78c96e05bae4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-24 12:24:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94468d75e20abe1b0583add763f833d791051f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:47:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325f38f37cce453891a89d2b0a383593 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a6a57fe061a4b3db8339851c0a1343f : SUCCESS in 19m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/08e26fdf7fd140ce9bf5d4498e27ed25 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cb56a78b6c12dec252be3dee293c74bed3daa806","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-24 13:47:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"759355cdae4fe38fcaaf9d8fc99b4a32545c4f42","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-24 15:08:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4eea2c56612231f8504b3c7d2d0afa96f417ef0a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-24 15:46:25.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nCould you please double-check some version bumps, see my comments inline.\n(Giving -1 to highlight my questions).","accounts_in_message":[],"_revision_number":2},{"id":"36e85aa15f61a247dc0efcf8991fe25238d84c08","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-24 15:46:29.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"8ad603057acd1b0384dafae2c06af0256174039c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 06:26:36.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8795da3527c4fe5557bf08574993f54ece363a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 06:26:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2fd3ae299429b83efc029586535ee31bc2ed620f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 06:42:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e6112944cda4bf698dfc25d2ba4ac80 : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9da4a1a91d94656bf7f59b45c16e168 : SUCCESS in 11m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0cf53241d5ea4d3cbb5fe2043c913cea : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"51e74454f60032e791de02916c2d8ec67bfee0fa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-26 08:21:37.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks yatin for the update \u0026 double-check! LGTM:\n* updated version bumps look OK to me\n* PTL is on vacation as mentioned in the comments so let\u0027s not wait for PTL-approved flag\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"b46952f1bbc016630011f83ea856c312a4390acc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-26 13:43:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c0a96e2908ec918bb8b4639618d7eb8a7a1d45e2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-27 15:10:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c620ef74e757ea6ce1daccc73f1ac3e19c15c64a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 15:10:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fff1d80c00f03fe113528cf0ee496f86a7f7c499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 15:23:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc87c25b4c8b4d989460c0ae9da074ed : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b895267173db46b9871c6cb4f288632b : FAILURE in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04e82f6cbb1a41589db9c11bf18aa61f : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7122d013a6818f5e782184d788d8816e9eeecdab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 06:56:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"73c0cfc18dd3e67b5bec627a6a47c3c127aa8c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:26:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39396481274f465c96a315d95e085173 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e29e86624a83442896193638f9af74fb : SUCCESS in 12m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/20104752ca54426791b16c1ee7051d95 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ee8d32204b43f084986b4260ef462df9be3f760c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:26:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"434edc4b248aeea45c941bc173142e2ebcc43ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:59:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a391fc8d7f74dccb38f9817334ff31e : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b9f9c6aa6b4bf8b777305db234937b : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3e21f91b07d4e8888fb5639189672f8 : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a49dd072d4b71e36fb7e469d42f90007c7e58beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0837d63a9b4d599e81f7966e7f101eec0ab1ec03","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-09-01 05:54:55.000000000","message":"Patch Set 3:\n\nthank you ykarel for updating this while i was out much appreciated 😊","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b3955b118a034ef30f71b2d30699d1b162417e4","revisions":{"c42d1ec6029a786e3284057d40c3c0f794edf0b6":{"kind":"REWORK","_number":1,"created":"2021-08-12 15:58:58.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/12/804412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/804412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/804412/1"}}},"commit":{"parents":[{"commit":"f3daa27bf77361c1363355ac062803ebe8b1cbf4","subject":"Merge \"Release OpenStack-Ansible Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3daa27bf77361c1363355ac062803ebe8b1cbf4"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-12 15:44:29.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-12 15:46:13.000000000","tz":180},"subject":"Make tripleo repos victoria release","message":"Make tripleo repos victoria release\n\nMake a tripleo repos release for the victoria branch. Depends-On\nis needed before validations pass as we tag puppet-tripleo here\nwith that commit.\n\nDepends-On: 564620e62f4f4943552dfd9f2ab645df29f9f198\nChange-Id: Ia1d8f84e3071bc24c6e059e4b3fcba2cc38601bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c42d1ec6029a786e3284057d40c3c0f794edf0b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c42d1ec6029a786e3284057d40c3c0f794edf0b6"}]},"branch":"refs/heads/master"},"f95aa19ef4803ffd9c5609a5ebc9a91d36830a6f":{"kind":"REWORK","_number":2,"created":"2021-08-24 12:22:39.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/804412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/804412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/804412/2"}}},"commit":{"parents":[{"commit":"0b385a7fafcff265af26212192b862cf9a2d30f4","subject":"Merge \"[oslo] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b385a7fafcff265af26212192b862cf9a2d30f4"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-12 15:44:29.000000000","tz":180},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-24 12:22:05.000000000","tz":330},"subject":"Make tripleo repos victoria release","message":"Make tripleo repos victoria release\n\nMake a tripleo repos release for the victoria branch. Depends-On\nis needed before validations pass as we tag puppet-tripleo here\nwith that commit.\n\nDepends-On: 564620e62f4f4943552dfd9f2ab645df29f9f198\nChange-Id: Ia1d8f84e3071bc24c6e059e4b3fcba2cc38601bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f95aa19ef4803ffd9c5609a5ebc9a91d36830a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f95aa19ef4803ffd9c5609a5ebc9a91d36830a6f"}]},"branch":"refs/heads/master"},"6b3955b118a034ef30f71b2d30699d1b162417e4":{"kind":"REWORK","_number":3,"created":"2021-08-26 06:26:37.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/804412/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/804412/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/804412/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/804412/3"}}},"commit":{"parents":[{"commit":"2e9a539c1dcffee2ce2c7bbd9d45703595c30c7e","subject":"Merge \"[winstackers] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e9a539c1dcffee2ce2c7bbd9d45703595c30c7e"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-12 15:44:29.000000000","tz":180},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-26 06:26:08.000000000","tz":330},"subject":"Make tripleo repos victoria release","message":"Make tripleo repos victoria release\n\nMake a tripleo repos release for the victoria branch. Depends-On\nis needed before validations pass as we tag puppet-tripleo here\nwith that commit.\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-tripleo/+/804411\nChange-Id: Ia1d8f84e3071bc24c6e059e4b3fcba2cc38601bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b3955b118a034ef30f71b2d30699d1b162417e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b3955b118a034ef30f71b2d30699d1b162417e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
