)]}'
{"id":"openstack%2Freleases~804562","triplet_id":"openstack%2Freleases~master~Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634","project":"openstack/releases","branch":"master","topic":"storlets-eol","hashtags":[],"change_id":"Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634","subject":"[storlets] EOL Pike, Queens, Rocky and Stein","status":"MERGED","created":"2021-08-14 12:51:29.000000000","updated":"2021-08-31 13:49:27.000000000","submitted":"2021-08-31 13:49:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804562-storlets-eol","meta_rev_id":"59120104111b953e29a1527ae0641c90cc796d9b","_number":804562,"virtual_id_number":804562,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 13:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-08-31 13:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-08-25 10:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-08-31 13:32:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-24 04:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"}],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-14 12:51:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2021-08-14 12:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 20:36:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-08-25 10:16:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-25 14:59:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"f88cc2e51a9cf8252257855cabb859674f2277c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 12:51:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5179d96ce4db1e4ea66a6a716a4a47a763002db","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 12:52:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c865fa37bf8b3acf04467cdf452be82aafedce50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 12:59:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef3a9e85f474f5bb3f0d2e1ec8b8f48 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ef84c7e78004badb94c2b8797af356e : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7fe4c0b67e4742c58d871883e8ce99e2 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4f316f27c5a8be9bd53ce384ede65a1fba196f8d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:55:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fd626664ac04bc9a5d9db4e7636fa7eaeef926e8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:56:38.000000000","message":"Patch Set 1: -Workflow\n\nI confirmed we made the first release in Queens and have no release in Pike. So starting from Queens is correct.","accounts_in_message":[],"_revision_number":1},{"id":"e18fd3d8c8faf799f34fa6bbfbc7adbe0b5816e7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 15:03:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"93fdc5b17c4a547821e32fd98595b4c094d17f4c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 20:36:40.000000000","message":"Patch Set 1:\n\nIs this waiting on anything Takashi?","accounts_in_message":[],"_revision_number":1},{"id":"07e685ce227e802c0b99e1fc99da555884b543a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-24 04:31:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b308b19cd508832592048c47b3cdcc4826dadf08","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-24 04:31:26.000000000","message":"Topic set to storlets-eol","accounts_in_message":[],"_revision_number":2},{"id":"7915d633a7c3243a7f945e852c04973b8b57047c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-24 04:31:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e104dc9469e04b8302c1b101613442b758896010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-24 04:32:30.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"aab242f95e78115f4dfc14ac13775ec6434bb946","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 04:33:38.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"f2133e76a306137d8fcd34bca738a360254a575f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-24 04:33:45.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is this waiting on anything Takashi?\n\nThis one is ready now. I\u0027ll submit a mail to openstack-discuss to notice these EOLs but I believe nobody would object.\nThe reason why I posted WIP was I notice the repo still has older branches like Pike, Ocata and Newton (!!!) but these are now all covered by this patch or the parent patch.","accounts_in_message":[],"_revision_number":4},{"id":"13792d4ecad623e45b3d2493e0172ea5bf957639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 04:42:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb9bc436ce446f8bbf28788f1ef8b01 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61e55e714e084b7ba4166d40a436687f : SUCCESS in 4m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/772e6ac7c9bb46d896cf756906167b23 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a9f7d0bdb70dbe1fd01b87fa855390311ef26089","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-25 10:16:46.000000000","message":"Patch Set 4: Code-Review+2\n\nUntil now nobody complained.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-August/024357.html","accounts_in_message":[],"_revision_number":4},{"id":"fda9cefdeb8d7955afcf87cc36c8059dea645780","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-25 14:59:44.000000000","message":"Patch Set 4: Code-Review+1\n\nhashes look good, let\u0027s wait some time as the mail was only sent yesterday to the ML.","accounts_in_message":[],"_revision_number":4},{"id":"152d40d8952502e28b37c15269041b64cc7b89bf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-31 13:32:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nOne week passed since Takashi sent the mail to ML. We have waited enough, so let\u0027s merge this.","accounts_in_message":[],"_revision_number":4},{"id":"2afb47b0cbcc90f1ca45ad0d1e990d93ccd01f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 13:32:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ba5512075fc33643e6bd2e7c428a8516162b8ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 13:49:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b94ed44997e2479a86ab16d34c2784f4 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa280c49b73642689ea5d2e08702dd4c : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5133403bbdc7450099aa0683621782f2 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"59120104111b953e29a1527ae0641c90cc796d9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 13:49:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e42755d9f96fd4de17a2587c7fa124476a7e72d6","revisions":{"20bd9e640cf01cbd83d528ee85fa12f26ae5aa1f":{"kind":"REWORK","_number":1,"created":"2021-08-14 12:51:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/804562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/804562/1"}}},"commit":{"parents":[{"commit":"f5a9c6c49f1c60512318b8bf62f408804fb3242b","subject":"Merge \"[heat] Propose to switch the release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5a9c6c49f1c60512318b8bf62f408804fb3242b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 12:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 12:50:44.000000000","tz":540},"subject":"Storlets: EOL Queens, Rocky and Stein","message":"Storlets: EOL Queens, Rocky and Stein\n\nThese three branches are quite old and we have seen no requirement\nor attempt of backports. Thus I\u0027d propose marking these branches as EOL\nto reduce number of branches maintained.\n\nChange-Id: Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20bd9e640cf01cbd83d528ee85fa12f26ae5aa1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20bd9e640cf01cbd83d528ee85fa12f26ae5aa1f"}]},"branch":"refs/heads/master"},"5a0ea37d191b8dc5d02441c3360fd20e09af038c":{"kind":"REWORK","_number":2,"created":"2021-08-24 04:31:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/804562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/804562/2"}}},"commit":{"parents":[{"commit":"f5a9c6c49f1c60512318b8bf62f408804fb3242b","subject":"Merge \"[heat] Propose to switch the release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5a9c6c49f1c60512318b8bf62f408804fb3242b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 12:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-24 04:30:55.000000000","tz":540},"subject":"Storlets: EOL Pike, Queens, Rocky and Stein","message":"Storlets: EOL Pike, Queens, Rocky and Stein\n\nThese four branches are quite old and we have seen no requirement\nor attempt of backports.\nThus I\u0027d propose marking these branches as EOL to reduce number of\nbranches maintained.\n\nChange-Id: Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a0ea37d191b8dc5d02441c3360fd20e09af038c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a0ea37d191b8dc5d02441c3360fd20e09af038c"}]},"branch":"refs/heads/master"},"6bc5a38634e00883ce82be5e287162d17d78b1ee":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-24 04:31:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/804562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/804562/3"}}},"commit":{"parents":[{"commit":"f5a9c6c49f1c60512318b8bf62f408804fb3242b","subject":"Merge \"[heat] Propose to switch the release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5a9c6c49f1c60512318b8bf62f408804fb3242b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 12:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-24 04:31:41.000000000","tz":540},"subject":"[storlets] EOL Pike, Queens, Rocky and Stein","message":"[storlets] EOL Pike, Queens, Rocky and Stein\n\nThese four branches are quite old and we have seen no requirement\nor attempt of backports.\nThus I\u0027d propose marking these branches as EOL to reduce number of\nbranches maintained.\n\nChange-Id: Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bc5a38634e00883ce82be5e287162d17d78b1ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bc5a38634e00883ce82be5e287162d17d78b1ee"}]},"branch":"refs/heads/master"},"e42755d9f96fd4de17a2587c7fa124476a7e72d6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-24 04:32:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804562/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/804562/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/804562/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/804562/4"}}},"commit":{"parents":[{"commit":"349828f39897bea36533cf2cf2b715e4cc06d022","subject":"[storlets] Transition Newton to Ocata to End of Life","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/349828f39897bea36533cf2cf2b715e4cc06d022"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 12:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-24 04:32:30.000000000","tz":0},"subject":"[storlets] EOL Pike, Queens, Rocky and Stein","message":"[storlets] EOL Pike, Queens, Rocky and Stein\n\nThese four branches are quite old and we have seen no requirement\nor attempt of backports.\nThus I\u0027d propose marking these branches as EOL to reduce number of\nbranches maintained.\n\nChange-Id: Idbb89fb5e9a9dc58fd7c9c2399ca4d13aa925634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e42755d9f96fd4de17a2587c7fa124476a7e72d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e42755d9f96fd4de17a2587c7fa124476a7e72d6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
