)]}'
{"id":"openstack%2Freleases~804659","triplet_id":"openstack%2Freleases~master~Ieb5b7c3e8bcf8a92a07b10cb11f78729ea94deaf","project":"openstack/releases","branch":"master","topic":"xena-final-non-client-libs","hashtags":[],"change_id":"Ieb5b7c3e8bcf8a92a07b10cb11f78729ea94deaf","subject":"Release final heat-translator for Xena","status":"MERGED","created":"2021-08-16 09:25:40.000000000","updated":"2021-08-20 13:55:15.000000000","submitted":"2021-08-20 13:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804659-xena-final-non-client-libs","meta_rev_id":"c02d27e560bb8549fbf1ece1195b5f786b5db899","_number":804659,"virtual_id_number":804659,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 13:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":1,"date":"2021-08-20 12:52:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2021-08-19 11:58:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-08-19 13:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2021-08-20 13:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-20 12:54:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 09:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-16 09:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 11:21:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-08-18 14:30:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-08-19 11:58:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"58d77d7c5c75b890c57340b3fb76978aa814aa87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 09:25:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53c793c14c6c3b0242be80dd332e2941c93bd491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edeee18f8d84172bf3f2319a454d52b : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a12ec867dd348039ccd1fe743c3d47e : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3279f07b3122488aa8742c96bf2ada6a : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cec83b456e1632c678e0731397b06f77401df61f","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-08-16 12:45:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad9c44a75f7c5109bcdeb155a8596f339f590740","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:46:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d03a28ea07fa9fd707e858fd172322e7e09d146d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-18 14:30:56.000000000","message":"Patch Set 1:\n\nWell, for me the MINOR version bump looks a bit overkill for the given release content (a single doc change). Especially as there was a release already for xena. So I\u0027d rather bumped the PATCH version only.","accounts_in_message":[],"_revision_number":1},{"id":"7a52324d1edc8ba587676e66c7103235596008be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 07:32:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"343f326310699d3fdad71337a1b38ad753820441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 07:42:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d54933d30594089a55915ad89c237e9 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/114822cce5524b5093333572b4734e2c : SUCCESS in 7m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0251675035f44c9584a6bb6bbaa8553f : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9b8275de97c078474fd8fa24770770f3c265e062","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-19 11:58:28.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update Hervé o:)","accounts_in_message":[],"_revision_number":2},{"id":"843eb9d40aa2f6c99d86cb7409fbe3896f08d74c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:32:45.000000000","message":"Patch Set 2: Code-Review+2\n\nWaiting for Rico\u0027s validation","accounts_in_message":[],"_revision_number":2},{"id":"dcafa9c2567f6931c99259083fb911b696a0f480","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-20 11:35:35.000000000","message":"Patch Set 2:\n\nRico please can you have another look to validate the latest changes and Elod\u0027s previous suggestion. Thanks by advance.","accounts_in_message":[],"_revision_number":2},{"id":"0c5af82acd06ede71d72880351d3598f25ea2ef2","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-08-20 12:52:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3e9848bd79495e7744eaa1a359044fc00a6031a4","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-08-20 12:53:46.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Rico please can you have another look to validate the latest changes and Elod\u0027s previous suggestion. Thanks by advance.\n\nMake sense to me, thanks @Hervé Beraud @Elod Illes","accounts_in_message":[],"_revision_number":2},{"id":"6bfbbd79e22ffbd79f632cf66ae8dc0af34316fd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 12:54:07.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"de50ac7d08e53e325fcb6f7c6e658e30493eab9f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-20 13:46:13.000000000","message":"Patch Set 2: Workflow+1\n\nApproved by PTL (thanks!)","accounts_in_message":[],"_revision_number":2},{"id":"05129a90ad9b1b2c35ebead250a85944c9d1ba60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 13:46:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec14c27a989bebd78a7a0bd02b287b624748ed69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 13:55:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f112576a004bf6955bfb9ef3c58018 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d465b1db85a141a591db82401ded3695 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/920efeeec3214d84b3f56421b67667d3 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c02d27e560bb8549fbf1ece1195b5f786b5db899","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 13:55:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fe5a4e15cce085f4ee873c60e55a80702aaec4d","revisions":{"c27b3dc31ac772d0caa71b0e36d8efde8387f76e":{"kind":"REWORK","_number":1,"created":"2021-08-16 09:25:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/804659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/804659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/804659/1"}}},"commit":{"parents":[{"commit":"39ef71ed213cb1a259d25de16631c28b8aa806fc","subject":"Merge \"Release OpenStack-Ansible Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39ef71ed213cb1a259d25de16631c28b8aa806fc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:25:32.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:25:32.000000000","tz":120},"subject":"Release final heat-translator for Xena","message":"Release final heat-translator for Xena\n\nThis proposed a release for heat-translator to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, August 23.\n\n$ git log --oneline --no-merges 2.4.0..0b8fba5\n0b8fba5 [ussuri][goal] Update contributor documentation\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ieb5b7c3e8bcf8a92a07b10cb11f78729ea94deaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c27b3dc31ac772d0caa71b0e36d8efde8387f76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c27b3dc31ac772d0caa71b0e36d8efde8387f76e"}]},"branch":"refs/heads/master"},"0fe5a4e15cce085f4ee873c60e55a80702aaec4d":{"kind":"REWORK","_number":2,"created":"2021-08-19 07:32:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/804659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/804659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/804659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/804659/2"}}},"commit":{"parents":[{"commit":"39ef71ed213cb1a259d25de16631c28b8aa806fc","subject":"Merge \"Release OpenStack-Ansible Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39ef71ed213cb1a259d25de16631c28b8aa806fc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:25:32.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-19 07:32:28.000000000","tz":120},"subject":"Release final heat-translator for Xena","message":"Release final heat-translator for Xena\n\nThis proposed a release for heat-translator to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, August 23.\n\n$ git log --oneline --no-merges 2.4.0..0b8fba5\n0b8fba5 [ussuri][goal] Update contributor documentation\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ieb5b7c3e8bcf8a92a07b10cb11f78729ea94deaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0fe5a4e15cce085f4ee873c60e55a80702aaec4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0fe5a4e15cce085f4ee873c60e55a80702aaec4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
