)]}'
{"id":"openstack%2Freleases~804676","triplet_id":"openstack%2Freleases~master~Icc048cb736d1e73737b98f6134e955d22bf46cf7","project":"openstack/releases","branch":"master","topic":"xena-final-non-client-libs","hashtags":[],"change_id":"Icc048cb736d1e73737b98f6134e955d22bf46cf7","subject":"Release final oslo.db for Xena","status":"MERGED","created":"2021-08-16 09:50:13.000000000","updated":"2021-08-23 10:20:06.000000000","submitted":"2021-08-23 10:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804676-xena-final-non-client-libs","meta_rev_id":"b945d43220b05dd84280053704601c8674545e24","_number":804676,"virtual_id_number":804676,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 10:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-08-23 10:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-23 09:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-08-23 10:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-08-20 11:33:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 09:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 09:52:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-08-19 07:56:51.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-08-19 12:10:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-08-19 13:00:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4ba413be9ef376eb9a01d6218f8dbb081c7aae13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 09:50:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee5bcf66210228b5a2d8d751c18d132516cc3970","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:51:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2fd1005227f9a589bf07000d6b0ca0b7ac736db3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 09:52:28.000000000","message":"Patch Set 1:\n\nHello Stephen,\n\nPlease, let me know if this version is ok for you or if you want including more patches. Also, please, let me know if the minor version is ok for you.\n\nThanks by advance","accounts_in_message":[],"_revision_number":1},{"id":"26d8773649bce3640337b966396321284e0df4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffedfce5b0c45b4911dd80e6fab7ce6 : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e68d10bce20c4566938b2fc183657e46 : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dcbe653381244e758393811850c6fc0e : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"62b6ca2c60c7bb252d6e0f7e1cc036ce3f657b3a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 11:50:30.000000000","message":"Patch Set 1: Code-Review-1\n\nHolding this patch until Stephen give feedbacks","accounts_in_message":[],"_revision_number":1},{"id":"043d8f9f24d29aaaf8e9b35cfd289db90b8bc038","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 14:07:40.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e Hello Stephen,\n\u003e \n\u003e Please, let me know if this version is ok for you or if you want including more patches. Also, please, let me know if the minor version is ok for you.\n\u003e \n\u003e Thanks by advance\n\nI think I would like to land more of the patches, but this hash will suffice if needed. When does this *have* to land by? Do we have the week?\n\nIn any case, this needs to be a major version since there are likely to be breaking changes in the SQLA 2.0 changes. We\u0027ve already seen some issues pop up with this in glance, as I\u0027ve noted on openstack-discuss. The -1 is for that","accounts_in_message":[],"_revision_number":1},{"id":"9b40cbc0470dfb2a6768f1dc77d6ae3f3347195e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 16:17:25.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Hello Stephen,\n\u003e \u003e \n\u003e \u003e Please, let me know if this version is ok for you or if you want including more patches. Also, please, let me know if the minor version is ok for you.\n\u003e \u003e \n\u003e \u003e Thanks by advance\n\u003e \n\u003e I think I would like to land more of the patches, but this hash will suffice if needed. When does this *have* to land by? Do we have the week?\n\nDeadline is Monday 23. After this point we will enter into oslo\u0027s feature freeze and the non client libs feature freeze.\n\n\u003e \n\u003e In any case, this needs to be a major version since there are likely to be breaking changes in the SQLA 2.0 changes. \n\nLGTM\n\nWe\u0027ve already seen some issues pop up with this in glance, as I\u0027ve noted on openstack-discuss. The -1 is for that\n\nAck","accounts_in_message":[],"_revision_number":1},{"id":"5b8db6d3d66941ef1fc45a988770deab4351d66e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 07:43:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0700f70930fc1e6f1e56fe36a9f71f0994ad4517","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 07:44:35.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"90e1105d0de92810b372eaa7377c3e4bf53e128c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 07:54:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc5fc794a5144aea592c975eccb514a : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97cd285fabf04ed3a2b33b0b47bb95c8 : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4a19448d69084faca9e851d29cb63f33 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9f9beb645c8a06780f0aa04cf745e4adc803c5ef","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-19 07:56:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"317545f88b81a0c7aaa1e9cb8149804daa460ac6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-19 12:10:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf2bfafe7a93c8c222f494645c1a421c5d5678e8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-19 13:00:31.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks OK to me, but do we want to wait to @stephenfin to review this again? Do you want to wait for another patches to be part of this release?","accounts_in_message":[],"_revision_number":2},{"id":"4acbf96eb01787e1c9c4449959fc6ce9de4bad77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-20 11:32:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18fc24935314d3f965b230a80fd9092f0d02b2ac","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 11:33:28.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"55632334ddd792c6790343e95865214b975066d0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-20 11:34:18.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e Looks OK to me, but do we want to wait to @stephenfin to review this again? Do you want to wait for another patches to be part of this release?\n\nI just rebased the hash to the HEAD and asked Stephen to do another review. Let\u0027s wait until this evening.","accounts_in_message":[],"_revision_number":3},{"id":"fcda7ffb3ff807ca7241b4d5d45016f9d9f3fc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 11:43:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd491f62d4b4785bbecd11501012e77 : SUCCESS in 8m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7411f18960a444f5bdd914332dd696c9 : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c73278edea6049b0a4098ac46aa58491 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":3},{"id":"33cdfa8991f55f8349f12bef5e8402229323c917","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-20 16:16:50.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review+1\n\u003e \u003e \n\u003e \u003e Looks OK to me, but do we want to wait to @stephenfin to review this again? Do you want to wait for another patches to be part of this release?\n\u003e \n\u003e I just rebased the hash to the HEAD and asked Stephen to do another review. Let\u0027s wait until this evening.\n\nI\u0027d like to get a few more patches in before we cut this, but the deadline is Monday and it doesn\u0027t seem likely to happen today. We can merge this as-is and cut a 12.0.0 release early in Yoga, I guess","accounts_in_message":[],"_revision_number":3},{"id":"a79f8723cde5fb1bd482e4bf20a4333e12b0ba1c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-23 09:17:41.000000000","message":"Patch Set 3: Code-Review+2\n\nOk thanks Stephen. Let\u0027s continue with that. We\u0027ll release a new major version early in Yoga.","accounts_in_message":[],"_revision_number":3},{"id":"8915dd6295e1828a13a441894a5054c184c0f664","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-23 10:01:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nWe can proceed with this patch, thanks for the reply stephenfin!\nLGTM:\n* MAJOR version bump looks OK\n* approved by the team\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"c7862320f202a0168dd15c455f27d1248bec9f67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:01:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2eec1d6412f32e0e63d79b5fa5b6abcdfc2bfa85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:20:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac2a093cafc44b45b95b3e19602ffae3 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0dc16e568ae495d8fe60568bd769cfc : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d63727813074bbba21e4a4a430f8535 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"b945d43220b05dd84280053704601c8674545e24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:20:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7484d9c84801751a9ffd2f47fd5752e4140f046a","revisions":{"6555eef7a5b7a59a5f3ade242ea9ed32e1970e00":{"kind":"REWORK","_number":1,"created":"2021-08-16 09:50:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/804676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/804676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/804676/1"}}},"commit":{"parents":[{"commit":"39ef71ed213cb1a259d25de16631c28b8aa806fc","subject":"Merge \"Release OpenStack-Ansible Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39ef71ed213cb1a259d25de16631c28b8aa806fc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:50:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:50:05.000000000","tz":120},"subject":"Release final oslo.db for Xena","message":"Release final oslo.db for Xena\n\nThis proposed a release for oslo.db to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, August 23.\n\n$ git log --oneline --no-merges 10.0.0..6dc4b1c\n6dc4b1c Replace use of insert.values parameter\nd949861 Add missing bind argument to calls\n872bc1d Don\u0027t pass strings to Connection.execute()\n22b44ee Remove use of MetaData.bind argument\n8d76072 Remove legacy calling style of select()\n4c1eb96 tests: Enable SQLAlchemy 2.0 deprecation warnings\nda002a8 utils: Deprecate sqlalchemy-migrate-related functions\n40bce5a tests: Enable SADeprecationWarning warnings\na6007a9 tests: Use common base class\n538e0a2 tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n73e376d tests: Clean up base test\nbd69e86 Drop checks for IBM DB2\nf048068 tox: Simplify test running\n4f9df08 tests: Remove \u0027ModelsMigrationsSync.check_foreign_keys\u0027\n74c6bf2 concurrency: Deprecate \u0027TpoolDbapiWrapper\u0027\na778f2d sqlalchemy: Remove checks for older deps\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Icc048cb736d1e73737b98f6134e955d22bf46cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6555eef7a5b7a59a5f3ade242ea9ed32e1970e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6555eef7a5b7a59a5f3ade242ea9ed32e1970e00"}]},"branch":"refs/heads/master"},"b3a82c9f9b6bf83f14536249d57251b22605958b":{"kind":"REWORK","_number":2,"created":"2021-08-19 07:43:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/804676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/804676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/804676/2"}}},"commit":{"parents":[{"commit":"39ef71ed213cb1a259d25de16631c28b8aa806fc","subject":"Merge \"Release OpenStack-Ansible Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39ef71ed213cb1a259d25de16631c28b8aa806fc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:50:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-19 07:42:26.000000000","tz":120},"subject":"Release final oslo.db for Xena","message":"Release final oslo.db for Xena\n\nThis proposed a release for oslo.db to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, August 23.\n\n$ git log --oneline --no-merges 10.0.0..8e7454c\n8e7454c Don\u0027t use plain string SQL statements\n6a013c6 Replace use of Table.autoload parameter\nd2f101c Replace use of update.whereclause parameter\n0816efc Update import of declarative_base()\n38c2eff Replace use of update.values parameter\n6dc4b1c Replace use of insert.values parameter\nd949861 Add missing bind argument to calls\n872bc1d Don\u0027t pass strings to Connection.execute()\n22b44ee Remove use of MetaData.bind argument\n8d76072 Remove legacy calling style of select()\n4c1eb96 tests: Enable SQLAlchemy 2.0 deprecation warnings\nda002a8 utils: Deprecate sqlalchemy-migrate-related functions\n40bce5a tests: Enable SADeprecationWarning warnings\na6007a9 tests: Use common base class\n538e0a2 tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n73e376d tests: Clean up base test\nbd69e86 Drop checks for IBM DB2\nf048068 tox: Simplify test running\n4f9df08 (HEAD -\u003e review/stephen_finucane/cleanup) tests: Remove\n\u0027ModelsMigrationsSync.check_foreign_keys\u0027\n74c6bf2 concurrency: Deprecate \u0027TpoolDbapiWrapper\u0027\na778f2d sqlalchemy: Remove checks for older deps\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Icc048cb736d1e73737b98f6134e955d22bf46cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3a82c9f9b6bf83f14536249d57251b22605958b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3a82c9f9b6bf83f14536249d57251b22605958b"}]},"branch":"refs/heads/master"},"7484d9c84801751a9ffd2f47fd5752e4140f046a":{"kind":"REWORK","_number":3,"created":"2021-08-20 11:32:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/804676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/804676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/804676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/804676/3"}}},"commit":{"parents":[{"commit":"39ef71ed213cb1a259d25de16631c28b8aa806fc","subject":"Merge \"Release OpenStack-Ansible Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39ef71ed213cb1a259d25de16631c28b8aa806fc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-16 09:50:05.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-20 11:31:51.000000000","tz":120},"subject":"Release final oslo.db for Xena","message":"Release final oslo.db for Xena\n\nThis proposed a release for oslo.db to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, August 23.\n\n$ git log --oneline --no-merges 10.0.0..75661fbc4d\nad4729d requirements: Bump sqlalchemy lower constraint\n8e7454c Don\u0027t use plain string SQL statements\n0816efc Update import of declarative_base()\n6a013c6 Replace use of Table.autoload parameter\n38c2eff Replace use of update.values parameter\nd2f101c Replace use of update.whereclause parameter\n6dc4b1c Replace use of insert.values parameter\nd949861 Add missing bind argument to calls\n872bc1d Don\u0027t pass strings to Connection.execute()\n22b44ee Remove use of MetaData.bind argument\n8d76072 Remove legacy calling style of select()\n4c1eb96 tests: Enable SQLAlchemy 2.0 deprecation warnings\nda002a8 utils: Deprecate sqlalchemy-migrate-related functions\n40bce5a tests: Enable SADeprecationWarning warnings\na6007a9 tests: Use common base class\n538e0a2 tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n73e376d tests: Clean up base test\nbd69e86 Drop checks for IBM DB2\nf048068 tox: Simplify test running\n4f9df08 (HEAD -\u003e review/stephen_finucane/cleanup) tests: Remove\n\u0027ModelsMigrationsSync.check_foreign_keys\u0027\n74c6bf2 concurrency: Deprecate \u0027TpoolDbapiWrapper\u0027\na778f2d sqlalchemy: Remove checks for older deps\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Icc048cb736d1e73737b98f6134e955d22bf46cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7484d9c84801751a9ffd2f47fd5752e4140f046a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7484d9c84801751a9ffd2f47fd5752e4140f046a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
