)]}'
{"id":"openstack%2Freleases~804871","triplet_id":"openstack%2Freleases~master~If33e654a43b077ea4c547d6302a8f7b2497251de","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"If33e654a43b077ea4c547d6302a8f7b2497251de","subject":"Move tripleo repos to independent release model","status":"MERGED","created":"2021-08-17 12:30:11.000000000","updated":"2021-08-19 07:06:33.000000000","submitted":"2021-08-19 07:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"804871","meta_rev_id":"d3aa69495f691cf69b60d05283386880ee3af25e","_number":804871,"virtual_id_number":804871,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 07:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":2,"date":"2021-08-18 14:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-08-18 18:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-08-19 06:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-08-19 06:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 12:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-17 14:20:08.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-08-17 14:37:26.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2021-08-17 14:56:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-17 15:33:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-08-18 10:03:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-08-18 13:18:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-18 14:59:41.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"1e0b4cae0ea1650d1a51a63c914642650dfe21da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 12:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55d45bb8cdeecfd0c4d47c3b4f526d4036c69c1c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 12:30:54.000000000","message":"Patch Set 1: Code-Review+1 Workflow-1\n\n+1 for PTL may be needed but workflow -1 until we are ready when the spec merges","accounts_in_message":[],"_revision_number":1},{"id":"69783967ced9d4ed787653477cac2fbcdc89b4da","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:31:29.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"63dc0246a2b4650da74f31a496540907c76dfe9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c82ed4ae2cd417daaa6ddd81a4872b9 : SUCCESS in 8m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3e5c1b0629642da921079c462d8ef94 : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a91eb4dc044e4a978d9ab4d33e8e513d : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"56c655ff9f56f19b8b78ce4ac3ede41a13f00617","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 14:05:17.000000000","message":"Patch Set 1: -Workflow\n\nthe spec has been worfklowed","accounts_in_message":[],"_revision_number":1},{"id":"eac73a9e4f75314cfc8f693012003c6ee6197057","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-17 14:20:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"71362974b40eadd569714bac099aa856511e4b01","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2021-08-17 14:37:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c9d85e2a63acb1bd6978804cec1c7eae5e71536","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-17 14:56:18.000000000","message":"Patch Set 1: Code-Review+1\n\nHello,\n\nAll your repos seems migrated:\n\n```\n$ tox -e venv -- list-deliverables --series xena --team tripleo\n...\nvenv run-test-pre: PYTHONHASHSEED\u003d\u00273794253545\u0027\nvenv run-test: commands[0] | list-deliverables --series xena --team tripleo\n_______________________________________________________ summary _______________________________________________________\n  venv: commands succeeded\n  congratulations :)\n```\n\nI think it\u0027s not too late to migrate these deliverables. We are close to the feature freeze and close to the milestone 3 but I think it is still ok to migrate them. However, before pushing the merging button I will discuss a bit about this with my team first. I only +1ed for now.","accounts_in_message":[],"_revision_number":1},{"id":"c0cbeafef218bb28202afb7a0935f9dde230a1e2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-17 15:33:40.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nStrictly from release yamls perspective, I have two comments/questions: I am unsure about puppet-tripleo and tripleo-ansible\u0027s release type change.\n\nOn the other hand, as I read through the linked mail thread and thinking on this change, I have multiple feelings. Anyway, clearly, I don\u0027t really have enough release manager experience to judge this change. I don\u0027t see what consequences it will have, what issues we may face. Though it could work I guess.","accounts_in_message":[],"_revision_number":1},{"id":"3bb4b9630e75e9de20e3feeb1e1072c02797e69e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 15:54:50.000000000","message":"Patch Set 1:\n\nthanks for comments Herve and Elod appreciate your thoughts as always\n\nfyi so you don\u0027t think i am ignoring you I will be out over the next few days but weshay will be watching here and can update if something is needed","accounts_in_message":[],"_revision_number":1},{"id":"458aeadd93b05faba2b2ff1736a62bb2b17c9a18","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 16:00:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a663c417cbbea36943ad9772bb222423df04945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-17 16:01:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"78bcbc504311c2f4ffb48f9b7fb32532ae532531","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:02:06.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8291215cd34f0ea50004c164666b166b50c070b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:09:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3739c238254d52aa390937101f602f : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1bd4c2cf06b4d0faf9ccd3735570dd0 : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/643f7a21158f4ee6be0130a4f33ccb85 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6d081d0100b83364bec92939c79bf45064566648","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-17 17:33:30.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Strictly from release yamls perspective, I have two comments/questions: I am unsure about puppet-tripleo and tripleo-ansible\u0027s release type change.\n\u003e \n\u003e On the other hand, as I read through the linked mail thread and thinking on this change, I have multiple feelings. Anyway, clearly, I don\u0027t really have enough release manager experience to judge this change. I don\u0027t see what consequences it will have, what issues we may face. Though it could work I guess.\n\nlooks like we do have artifacts in pypi. sorry I didn\u0027t realize that.","accounts_in_message":[],"_revision_number":2},{"id":"390465488309725d962a2692305f37ecc46da0fe","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-17 17:33:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b44ec714f368fe1d76988b26b9212744e9221592","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-17 17:36:12.000000000","message":"Patch Set 2:\n\nTo be clear.. as a project the only install method that is community supported is via rpm.\nThe pypi artifacts look like something that is just for openstack compliance.  Nothing *needs* to be in pypi to enable the community.\n\nhttps://docs.openstack.org/tripleo-docs/latest/\nhttps://www.rdoproject.org/","accounts_in_message":[],"_revision_number":2},{"id":"7444ef9115e6817b0787973c7d494de4fc91c7d1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-18 10:03:53.000000000","message":"Patch Set 2:\n\nThis looks good to me, my only question would be around the need to release tripleo-ansible on Pypi. It\u0027s never been released there, so I\u0027m not sure we should start now? It would only be required for \"OpenStack compliance\" if we consumed those in our requirements, and that is not the case.","accounts_in_message":[],"_revision_number":2},{"id":"94b9fdf9524989283616ff7c570a2008cb1ee3dc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-18 10:06:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a70d00f03dc6c82fc50ac56cd44fcd6f14703021","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-18 13:18:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f2d883162b18de8a5379cace95ccb3bed710a58","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-18 13:19:40.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b012700ea59690ee7bdd2eff95977aaaeca875a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-18 13:45:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2b2723550fe740e6314818db867eae84489e14e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 13:58:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0a2ca091b54706b7b1f9f3381635cb : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89335ecbf2664a3aade918013ce0c121 : SUCCESS in 6m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a5bc6324b0ba43e39e321c5fad2d37f1 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"6f33977da578b39970ead3bdc96a76e98169a3ef","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-18 14:59:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e3c76576c9259c10a67b922f2627e21a14d90b53","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-18 18:02:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d0b8dc29fab07ae8c68bb5ee8b7ae385b8354ac0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 06:56:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fd837d176667f84f96d67e712e87b7901df1cdf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 06:57:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eef64b5cb992a6fd2b81bad0cfae2a12c64847a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 07:06:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd67bd693824cb39d8f73e4e2a513e0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4d1ace51474526ad9e522b8d7f92bf : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8c7c59ace4140eb8ed8f9a851231268 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d3aa69495f691cf69b60d05283386880ee3af25e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 07:06:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"230cb14c4a0fa8f3819ed24776497448dca122a8","revisions":{"66d003c39d5d62cd7721a64ca39c34827cfe60da":{"kind":"REWORK","_number":1,"created":"2021-08-17 12:30:11.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/71/804871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/804871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/804871/1"}}},"commit":{"parents":[{"commit":"c44b496e81eb17b9a09604d941d88412fe7e5a07","subject":"Merge \"Remove devstack-gate feature list related task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c44b496e81eb17b9a09604d941d88412fe7e5a07"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-17 12:14:13.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-17 12:14:13.000000000","tz":180},"subject":"Move tripleo repos to independent release model","message":"Move tripleo repos to independent release model\n\nAs discussed at [1] this moves all tripleo repos to the _independent\nfolder and removes the xena one.\n\n[1] https://review.opendev.org/c/openstack/tripleo-specs/+/801512\n\nChange-Id: If33e654a43b077ea4c547d6302a8f7b2497251de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66d003c39d5d62cd7721a64ca39c34827cfe60da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66d003c39d5d62cd7721a64ca39c34827cfe60da"}]},"branch":"refs/heads/master"},"266bb1c232b6169dd61467dda3b4aba873cceae4":{"kind":"REWORK","_number":2,"created":"2021-08-17 16:01:00.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/71/804871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/804871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/804871/2"}}},"commit":{"parents":[{"commit":"c44b496e81eb17b9a09604d941d88412fe7e5a07","subject":"Merge \"Remove devstack-gate feature list related task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c44b496e81eb17b9a09604d941d88412fe7e5a07"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-17 12:14:13.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-17 15:57:47.000000000","tz":0},"subject":"Move tripleo repos to independent release model","message":"Move tripleo repos to independent release model\n\nAs discussed at [1] this moves all tripleo repos to the _independent\nfolder and removes the xena one.\n\n[1] https://review.opendev.org/c/openstack/tripleo-specs/+/801512\n\nChange-Id: If33e654a43b077ea4c547d6302a8f7b2497251de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/266bb1c232b6169dd61467dda3b4aba873cceae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/266bb1c232b6169dd61467dda3b4aba873cceae4"}]},"branch":"refs/heads/master"},"230cb14c4a0fa8f3819ed24776497448dca122a8":{"kind":"REWORK","_number":3,"created":"2021-08-18 13:45:04.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/71/804871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/804871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/804871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/804871/3"}}},"commit":{"parents":[{"commit":"c44b496e81eb17b9a09604d941d88412fe7e5a07","subject":"Merge \"Remove devstack-gate feature list related task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c44b496e81eb17b9a09604d941d88412fe7e5a07"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-17 12:14:13.000000000","tz":180},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-08-18 13:44:05.000000000","tz":0},"subject":"Move tripleo repos to independent release model","message":"Move tripleo repos to independent release model\n\nAs discussed at [1] this moves all tripleo repos to the _independent\nfolder and removes the xena one.\n\n[1] https://review.opendev.org/c/openstack/tripleo-specs/+/801512\n\nChange-Id: If33e654a43b077ea4c547d6302a8f7b2497251de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/230cb14c4a0fa8f3819ed24776497448dca122a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/230cb14c4a0fa8f3819ed24776497448dca122a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
