)]}'
{"id":"openstack%2Freleases~806580","triplet_id":"openstack%2Freleases~master~Id3d28c9b49ccc5e08d36ce157d7acda5ae799be8","project":"openstack/releases","branch":"master","topic":"xena-milestone-3","hashtags":[],"change_id":"Id3d28c9b49ccc5e08d36ce157d7acda5ae799be8","subject":"Release final osc-placement for Xena","status":"MERGED","created":"2021-08-30 13:47:04.000000000","updated":"2021-09-03 10:24:46.000000000","submitted":"2021-09-03 10:24:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806580-xena-milestone-3","meta_rev_id":"05449bd23f96f69425321427397b23811adc8fbf","_number":806580,"virtual_id_number":806580,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 10:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-03 10:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-03 07:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-03 07:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-09-03 10:03:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-03 07:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 13:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 13:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 11:21:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-31 11:21:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-09-03 10:03:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"dbaf252c35a43f0b707af7aba410c247064abeaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 13:47:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f82317326f2aef88e7e9985e3a6bd2005761795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 13:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c0a7bc23b8407c8127ad7a8e08b6fd : SUCCESS in 8m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65948d0a3e0a4c5595c64e17473dfac1 : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ffd1b898333b40e08fd55e07587daab9 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"04d11a437911828dc0313afd91829802c63739f5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 08:03:22.000000000","message":"Patch Set 1:\n\nHello,\n\nDeadline is today, please can you have a look to these changes and tell us if you are ready or not.\n\nThanks by advance.","accounts_in_message":[],"_revision_number":1},{"id":"01e5388bd1ea8356e297f5e44c9a497711d19595","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-02 14:28:53.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e Hello,\n\u003e \n\u003e Deadline is today, please can you have a look to these changes and tell us if you are ready or not.\n\u003e \n\u003e Thanks by advance.\n\nI\u0027d like to wait until https://review.opendev.org/c/openstack/osc-placement/+/804458 lands. It is on the gate now.","accounts_in_message":[],"_revision_number":1},{"id":"a58df9aa66e434490098101b14f4ebc1cefb94e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-02 14:35:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"68d5e4cc856ded1850a36cae38093fa75c26e7c8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-02 14:36:12.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Hello,\n\u003e \u003e \n\u003e \u003e Deadline is today, please can you have a look to these changes and tell us if you are ready or not.\n\u003e \u003e \n\u003e \u003e Thanks by advance.\n\u003e \n\u003e I\u0027d like to wait until https://review.opendev.org/c/openstack/osc-placement/+/804458 lands. It is on the gate now.\n\n^^ landed now.","accounts_in_message":[],"_revision_number":2},{"id":"6a2d3ce897b38bba0a3b3ce14241724a6b6f3074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 14:37:16.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3279b3ea3631c84f80cd5d3ab8622ae9b02fec89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 14:41:37.000000000","message":"Patch Set 3:\n\nYou beat me :) \nPS2 and PS3 are more or less identical.","accounts_in_message":[],"_revision_number":3},{"id":"eb251a9e4ce472ff659dc71fe5d317d880473a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:11:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4334370a41e24285a22cc780564c4c9e : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c54f72a132254383b3c3429c74b6f30c : SUCCESS in 7m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a24b2cb1cf45459d8596d80ed33e04bd : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5d4e17f49a8f636fd1f99997a80aa2d5e7210679","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 07:34:28.000000000","message":"Patch Set 3: Code-Review+2\n\nI think we can consider these changes as PTL approved as the PS2 and PS3 are identical.","accounts_in_message":[],"_revision_number":3},{"id":"424a1ea129adbb04bd751cb1d20488c438b04ba4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-03 07:35:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f5470c40d8fe03c96bc5f3844515a570ebfb2730","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:35:48.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"7ce8d2f5f68711ce6a1bcccdd7d63cc669a8da48","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-03 10:03:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM:\n* MINOR version bump looks OK\n* approved by PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"78c8036454789b1c9cc20966ed25d09d91de1bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:03:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cad1d58fe346f23ead415f671f781f00ae750d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:24:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2248d150d047579ac457e811abbf54 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6776a11eabaa4a31af4773966d65de0e : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0a5b6f8cbda44de84d10c7cd11fe3fe : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"05449bd23f96f69425321427397b23811adc8fbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:24:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d7d1270e7bd0722e9dd03bbbcdb52a364e2d88a","revisions":{"d19fe000b230305817a0741cf43b7d83675f4d73":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:47:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/80/806580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/806580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/806580/1"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:46:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:46:56.000000000","tz":120},"subject":"Release final osc-placement for Xena","message":"Release final osc-placement for Xena\n\nThis proposed a release for osc-placement to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 3.0.1..b0a74d0\n8b3ed3c Mark microversion 1.37 supported\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Id3d28c9b49ccc5e08d36ce157d7acda5ae799be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d19fe000b230305817a0741cf43b7d83675f4d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d19fe000b230305817a0741cf43b7d83675f4d73"}]},"branch":"refs/heads/master"},"00ffd778d4d395b3a68d225752045c30e6f102d3":{"kind":"REWORK","_number":2,"created":"2021-09-02 14:35:53.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/80/806580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/806580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/806580/2"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:46:56.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-02 14:35:45.000000000","tz":0},"subject":"Release final osc-placement for Xena","message":"Release final osc-placement for Xena\n\nThis proposed a release for osc-placement to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 3.0.1..5e38a59\n5e38a59 Add support for microversion 1.38 consumer types\n8b3ed3c Mark microversion 1.37 supported\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Id3d28c9b49ccc5e08d36ce157d7acda5ae799be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00ffd778d4d395b3a68d225752045c30e6f102d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00ffd778d4d395b3a68d225752045c30e6f102d3"}]},"branch":"refs/heads/master"},"6d7d1270e7bd0722e9dd03bbbcdb52a364e2d88a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-02 14:37:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/80/806580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/806580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/806580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/806580/3"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:46:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-02 14:36:03.000000000","tz":120},"subject":"Release final osc-placement for Xena","message":"Release final osc-placement for Xena\n\nThis proposed a release for osc-placement to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 3.0.1..HEAD\n5e38a59 Add support for microversion 1.38 consumer types\n8b3ed3c Mark microversion 1.37 supported\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Id3d28c9b49ccc5e08d36ce157d7acda5ae799be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d7d1270e7bd0722e9dd03bbbcdb52a364e2d88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d7d1270e7bd0722e9dd03bbbcdb52a364e2d88a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
