)]}'
{"id":"openstack%2Freleases~806589","triplet_id":"openstack%2Freleases~master~I8d585b46c92f835298ba1d3d1d8cc6a17d361af1","project":"openstack/releases","branch":"master","topic":"xena-milestone-3","hashtags":[],"change_id":"I8d585b46c92f835298ba1d3d1d8cc6a17d361af1","subject":"Release final python-manilaclient for Xena","status":"MERGED","created":"2021-08-30 13:57:52.000000000","updated":"2021-09-03 13:31:00.000000000","submitted":"2021-09-03 13:31:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806589-xena-milestone-3","meta_rev_id":"971a2a165a3cb12fbd3ba522823f9f152c6c4b14","_number":806589,"virtual_id_number":806589,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 13:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-09-03 12:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-03 13:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-03 12:08:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-03 13:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-03 12:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 14:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 11:21:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-08-31 11:21:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-09-03 09:56:01.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-09-03 12:47:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1c223e39326645666c2e4dd158d6a1f2978c38ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 13:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95a7575ff0044f83a2e3b2e64d56e75fc6f51188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 14:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2caf25824e5743d69121519b9a879af5 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa194468ab3d43f4a1cac2adad032a71 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1c390d1388624e4ea0f17304cfc2c4dd : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ada9e6e2a440a2602ab623abe3792df2cc46dd2d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 08:04:25.000000000","message":"Patch Set 1:\n\nHello,\n\nDeadline is today, please can you have a look to these changes and telling us if you are ready or not.\n\nThanks by advance.","accounts_in_message":[],"_revision_number":1},{"id":"d4b39a1ddb6f597aa1ae2c5a85e4dc3975833d77","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-02 17:57:39.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e Hello,\n\u003e \n\u003e Deadline is today, please can you have a look to these changes and telling us if you are ready or not.\n\u003e \n\u003e Thanks by advance.\n\nHi Herve, \n\nWe have a few changes on their final leg of merging; so can we hold off this patch? I\u0027m keeping tab of these changes and will post an update here soon!","accounts_in_message":[],"_revision_number":1},{"id":"6f81f5632e72af577080b18ba8a36f8e5cd6e31c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 18:30:12.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Hello,\n\u003e \u003e \n\u003e \u003e Deadline is today, please can you have a look to these changes and telling us if you are ready or not.\n\u003e \u003e \n\u003e \u003e Thanks by advance.\n\u003e \n\u003e Hi Herve, \n\u003e \n\u003e We have a few changes on their final leg of merging; so can we hold off this patch? I\u0027m keeping tab of these changes and will post an update here soon!\n\nSure no problem.","accounts_in_message":[],"_revision_number":1},{"id":"72de996881ae05275d01a59580e5643470ef8ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 03:11:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6599be7aca9c0733f71a9d5fd000e1e62b1c29f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 03:13:04.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9e0553d03f9bbaf4dc9f34b5a258e5552057a863","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 03:14:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hello,\n\u003e \u003e \u003e \n\u003e \u003e \u003e Deadline is today, please can you have a look to these changes and telling us if you are ready or not.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Thanks by advance.\n\u003e \u003e \n\u003e \u003e Hi Herve, \n\u003e \u003e \n\u003e \u003e We have a few changes on their final leg of merging; so can we hold off this patch? I\u0027m keeping tab of these changes and will post an update here soon!\n\u003e \n\u003e Sure no problem.\n\nThanks Hervé - we\u0027re ready now.","accounts_in_message":[],"_revision_number":2},{"id":"d33f73ea8c5684a6132b998e6b91bf208a972277","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 03:14:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bdcf89cd533ad8b1af3de80c98a145b9f207a15e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 03:32:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb4aa05ab34419b8c018e45a21f9143 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e56a556707649308f99575de5e41d43 : SUCCESS in 9m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6681b1a9754e4a8497711145fc0f6486 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b5a87a7b65f2294c2eacd998c4a842a4ce9fd8b5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 05:52:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b618bc760476a0c8b4abcc8f427fe05bc96d2e70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-03 09:56:01.000000000","message":"Patch Set 2:\n\nreno suggests MAJOR version bump [1] due to removal of deprecated features [2]. I think the version should be 3.0.0\n\n[1] https://zuul.opendev.org/t/openstack/build/6681b1a9754e4a8497711145fc0f6486/log/tox/list-changes-results.log#123\n[2] https://zuul.opendev.org/t/openstack/build/6681b1a9754e4a8497711145fc0f6486/log/tox/list-changes-results.log#979-987","accounts_in_message":[],"_revision_number":2},{"id":"63c303ab9ec488a1bb8be8749a4a1013cb012716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 11:19:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cec1cf085fdc5f9df0e61983e38fe551ae82e6c4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 11:20:11.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e reno suggests MAJOR version bump [1] due to removal of deprecated features [2]. I think the version should be 3.0.0\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/6681b1a9754e4a8497711145fc0f6486/log/tox/list-changes-results.log#123\n\u003e [2] https://zuul.opendev.org/t/openstack/build/6681b1a9754e4a8497711145fc0f6486/log/tox/list-changes-results.log#979-987\n\nAh yes good catch","accounts_in_message":[],"_revision_number":3},{"id":"380a5a082eb9357c1513dc87916e0d07f54c8483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:34:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679c8b0586b24852bf550aee836ec6b0 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1f78a23cf414eb492a30843cba510d6 : SUCCESS in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4d740dc802e74795906a335c0ae3c74a : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b7db36a85da43be7f8d885d5b64b49afbac402b0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-09-03 12:08:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1988a13356a8d3704bbb2175b14bd757aff0eb70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 12:12:23.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"cea8eb5505c85ea51ef4cad43f95b423c199929c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-03 12:47:30.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the update. LGTM. (+ approved by release liaison and no issues found in test job logs)","accounts_in_message":[],"_revision_number":3},{"id":"d3864fd0aa0701c7c9bd7314b08e3421b0c6d0e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 13:22:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"52f5fb1dd731b85c18bb304aed4abf470640859d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:22:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"565512a4737d36daeb4e399cb1f9531f50d99e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:31:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476b5796ce6f47549f8b0c0fbecce4e3 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5ef11a98ae44a688d4a842e242ed41 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/802a788a1f0d4e3a8f65fd7e2f51e086 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"971a2a165a3cb12fbd3ba522823f9f152c6c4b14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:31:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b882293ee47ce31569adc9ef394d88f1e02b25f","revisions":{"3e22084574ff12ff45a3ca474e98d79406bd2003":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:57:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/806589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/806589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/806589/1"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:57:45.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:57:45.000000000","tz":120},"subject":"Release final python-manilaclient for Xena","message":"Release final python-manilaclient for Xena\n\nThis proposed a release for python-manilaclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 2.7.0..2f173cb\n2f173cb [OSC] Implement share instance commands\n1d59127 [OSC] Add Share Snapshot Instances commands\n2fc21c3 Fix incorrect dict member lookup\nd566ecc [OSC] Implement Share Pools List command\n376429e Add --wait flag to the create share group operation\nca00798 [OSC] Implement Replica Export Locations commands\nfcbcf75 [OSC] Implement Share Services Commands\n53275be [OSC] Implement Availability Zones Command\n616f3ac Fix issue while creating share from snapshot\n3c278d7 Fix PDF docs issue\n6ecdbef Re-enable shell completion cache\nadc5bc1 [OSC] Fix the default API_VERSION in unit tests\nc5e2561 Replace deprecated inspect.getargspec\nf84f32e Add --wait flag to the delete share group operation\neea1d21 tests: Remove unused fakes\n0002625 utils: Remove unicode-related helpers\nf6c747f tools: Remove dead tools\n1562ec6 Remove six\nf3d7927 Uncap PrettyTable\n2a1fb2b Add \"--wait\" option for deleting a share server\n8ad1583 Fix typo in quota set error message\ndf33886 [OSC] Implement Share Replica Commands\n5220acf Changed minversion in tox to 3.18.0\nde19027 Further fix to \"test_list_shares_by_status\"\n931743a Remove some deprecated options\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I8d585b46c92f835298ba1d3d1d8cc6a17d361af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e22084574ff12ff45a3ca474e98d79406bd2003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e22084574ff12ff45a3ca474e98d79406bd2003"}]},"branch":"refs/heads/master"},"22b40f9c1340cf821d79385e93446c30affa88ef":{"kind":"REWORK","_number":2,"created":"2021-09-03 03:11:51.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/806589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/806589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/806589/2"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:57:45.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-03 03:11:32.000000000","tz":-420},"subject":"Release final python-manilaclient for Xena","message":"Release final python-manilaclient for Xena\n\n% git log --oneline --no-merges 2.7.0..89f7f27\n89f7f2728 Fix OSC share groups\n7c8aa8efd [OSC] Implement Share Group Commands\n92ca00cd6 Implement OSC share instance export location commands\nd40205a2f OSC Implementation for Share Networks\ncc440609b [OSC] Implement Share Limits Show command\n2f173cb15 [OSC] Implement share instance commands\n1d591270b [OSC] Add Share Snapshot Instances commands\n2fc21c3e0 Fix incorrect dict member lookup\nd566ecc09 [OSC] Implement Share Pools List command\n376429ec1 Add --wait flag to the create share group operation\nca0079803 [OSC] Implement Replica Export Locations commands\nfcbcf7521 [OSC] Implement Share Services Commands\n53275bec6 [OSC] Implement Availability Zones Command\n616f3acf6 Fix issue while creating share from snapshot\n3c278d7a0 Fix PDF docs issue\n6ecdbef0f Re-enable shell completion cache\nadc5bc193 [OSC] Fix the default API_VERSION in unit tests\nc5e2561ce Replace deprecated inspect.getargspec\nf84f32e70 Add --wait flag to the delete share group operation\neea1d21b8 tests: Remove unused fakes\n0002625e2 utils: Remove unicode-related helpers\nf6c747f15 tools: Remove dead tools\n1562ec6fb Remove six\nf3d792768 Uncap PrettyTable\n2a1fb2bf8 Add \"--wait\" option for deleting a share server\n8ad158327 Fix typo in quota set error message\ndf3388677 [OSC] Implement Share Replica Commands\n5220acf12 Changed minversion in tox to 3.18.0\nde1902786 Further fix to \"test_list_shares_by_status\"\n931743acb Remove some deprecated options\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I8d585b46c92f835298ba1d3d1d8cc6a17d361af1\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22b40f9c1340cf821d79385e93446c30affa88ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22b40f9c1340cf821d79385e93446c30affa88ef"}]},"branch":"refs/heads/master"},"5b882293ee47ce31569adc9ef394d88f1e02b25f":{"kind":"REWORK","_number":3,"created":"2021-09-03 11:19:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/806589/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/806589/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/806589/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/806589/3"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 13:57:45.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-03 11:19:40.000000000","tz":120},"subject":"Release final python-manilaclient for Xena","message":"Release final python-manilaclient for Xena\n\n% git log --oneline --no-merges 2.7.0..89f7f27\n89f7f2728 Fix OSC share groups\n7c8aa8efd [OSC] Implement Share Group Commands\n92ca00cd6 Implement OSC share instance export location commands\nd40205a2f OSC Implementation for Share Networks\ncc440609b [OSC] Implement Share Limits Show command\n2f173cb15 [OSC] Implement share instance commands\n1d591270b [OSC] Add Share Snapshot Instances commands\n2fc21c3e0 Fix incorrect dict member lookup\nd566ecc09 [OSC] Implement Share Pools List command\n376429ec1 Add --wait flag to the create share group operation\nca0079803 [OSC] Implement Replica Export Locations commands\nfcbcf7521 [OSC] Implement Share Services Commands\n53275bec6 [OSC] Implement Availability Zones Command\n616f3acf6 Fix issue while creating share from snapshot\n3c278d7a0 Fix PDF docs issue\n6ecdbef0f Re-enable shell completion cache\nadc5bc193 [OSC] Fix the default API_VERSION in unit tests\nc5e2561ce Replace deprecated inspect.getargspec\nf84f32e70 Add --wait flag to the delete share group operation\neea1d21b8 tests: Remove unused fakes\n0002625e2 utils: Remove unicode-related helpers\nf6c747f15 tools: Remove dead tools\n1562ec6fb Remove six\nf3d792768 Uncap PrettyTable\n2a1fb2bf8 Add \"--wait\" option for deleting a share server\n8ad158327 Fix typo in quota set error message\ndf3388677 [OSC] Implement Share Replica Commands\n5220acf12 Changed minversion in tox to 3.18.0\nde1902786 Further fix to \"test_list_shares_by_status\"\n931743acb Remove some deprecated options\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I8d585b46c92f835298ba1d3d1d8cc6a17d361af1\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5b882293ee47ce31569adc9ef394d88f1e02b25f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5b882293ee47ce31569adc9ef394d88f1e02b25f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
