)]}'
{"id":"openstack%2Freleases~806593","triplet_id":"openstack%2Freleases~master~I82ebacef170592867182de6ecf7e8176447cdaa6","project":"openstack/releases","branch":"master","topic":"xena-milestone-3","hashtags":[],"change_id":"I82ebacef170592867182de6ecf7e8176447cdaa6","subject":"Release final python-tackerclient for Xena","status":"MERGED","created":"2021-08-30 14:04:47.000000000","updated":"2021-09-03 11:25:31.000000000","submitted":"2021-09-03 11:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"806593-xena-milestone-3","meta_rev_id":"ee1656da8fdc09122e01843544f5098619a6ef48","_number":806593,"virtual_id_number":806593,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 11:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2021-09-03 10:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2021-09-03 10:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-03 11:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-03 11:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-03 10:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:06:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 14:17:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 11:20:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2021-09-02 08:57:22.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2021-09-03 10:21:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"bbe269805641531d4e5a2e902b2765bb1713f0e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 14:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b5626ebfe2137ced3065345074a6ce7d053b653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 14:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a593541dcc456aaa37574d2f6602a7 : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57abc4e506e548088d4a43db0ec0f5eb : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ca24b077a7664af8838bef1f6b89eaa5 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7d04bfd0bdc7f564551f53763d5923f03b77634f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 08:17:30.000000000","message":"Patch Set 1:\n\nHello,\n\nDeadline is today, please can you have a look to these changes and telling us if you are ready or not.\n\nThanks by advance.","accounts_in_message":[],"_revision_number":1},{"id":"12dbf40ca9aa7126888ceb9a8240829de142c454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-03 06:11:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7a460cf7c7863a113358ceca0d3e415d47667c4f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:27:45.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"689de8edc531eb9d535ea828efb2d42715318c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:44:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efebc5f3b77c496f8d521ec6a72e94fb : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb32dcc65bde4eeead1d374e6b3fe0f1 : SUCCESS in 6m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c3692217f8054ba495aaf6bacf3e9bf4 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"18cdf210ac66b8406a378054019dab5adcd726c8","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-03 06:51:17.000000000","message":"Patch Set 2: Code-Review+1\n\nSorry for slow reply. I\u0027ve fixed the update. Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"a6362a650e1a1681c10163407376c9aa3f8ef05d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 07:33:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nNo problem. Thank you.\nHowever I think we have an issue with your latest patch set, please see my inline comment (-1 related).","accounts_in_message":[],"_revision_number":2},{"id":"88400d780f861dec2cffedf99ba50952b1b08a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 09:11:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bab62151910ff1b4333ded3b4ea01b26d869424","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 09:12:13.000000000","message":"Patch Set 3:\n\nPS3 fixed the hash","accounts_in_message":[],"_revision_number":3},{"id":"a28b36e8f7c7291b540c22d42625029b373b6331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 09:24:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573380b8619047e4863063991def8421 : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9779fbd343043f1831102a56a722179 : SUCCESS in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0596f7318eb34a26a001230c820d05eb : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"45307afc052a2477d74971656863e364c897a39b","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-03 10:09:59.000000000","message":"Patch Set 3: Code-Review+1\n\nOops, thanks for correcting my mistake.","accounts_in_message":[],"_revision_number":3},{"id":"de91967cf62b848777ab973b1f3006ef393779d7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:11:35.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"d9b2b8b00804a41dd7c94f125b554554abf152b3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-03 10:21:58.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM:\n* MINOR version bump looks OK\n* approved by PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"93e1050804dd6b7be1a97bc81948c975f975a63a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 11:16:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e Oops, thanks for correcting my mistake.\n\nno problem you are welcome","accounts_in_message":[],"_revision_number":3},{"id":"4dfc9dd5ba189d3449c21cebc50b0771a707d2e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:17:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f42b440decc0a7e8d5d53cf20e4f19980d2b88f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:25:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d428ba924a84ac79a76982d977bd160 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78690a332b884421870174f3746c3371 : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a5e39b0c9f64263a25bacbb0519d47e : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"ee1656da8fdc09122e01843544f5098619a6ef48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:25:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75c7d75514bf78474e781082ba8c468fc9653a3c","revisions":{"b4a21ee4b3099d828233a0d7b2cd4132a62c72e5":{"kind":"REWORK","_number":1,"created":"2021-08-30 14:04:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/806593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/806593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/806593/1"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:04:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:04:22.000000000","tz":120},"subject":"Release final python-tackerclient for Xena","message":"Release final python-tackerclient for Xena\n\nThis proposed a release for python-tackerclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 1.7.0..6325127\n6325127 Remove use of testtools.helpers.safe_hasattr\ne8280f5 Remove broken link from Python-TackerClient guide\n94e28fc Correct YAML load warning\nc20ef24 vnfConfigurableProperties missing in VNF show\n4c59029 Update min version of tox\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I82ebacef170592867182de6ecf7e8176447cdaa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4a21ee4b3099d828233a0d7b2cd4132a62c72e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4a21ee4b3099d828233a0d7b2cd4132a62c72e5"}]},"branch":"refs/heads/master"},"1760bde16b79ee9bc9263a99a7830961f845f516":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-03 06:11:05.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/93/806593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/806593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/806593/2"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:04:22.000000000","tz":120},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-03 06:11:02.000000000","tz":0},"subject":"Release final python-tackerclient for Xena","message":"Release final python-tackerclient for Xena\n\nThis proposed a release for python-tackerclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 1.7.0..98789f8\n98789f8 Multi version API support\n6325127 Remove use of testtools.helpers.safe_hasattr\ne8280f5 Remove broken link from Python-TackerClient guide\n94e28fc Correct YAML load warning\nc20ef24 vnfConfigurableProperties missing in VNF show\n4c59029 Update min version of tox\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I82ebacef170592867182de6ecf7e8176447cdaa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1760bde16b79ee9bc9263a99a7830961f845f516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1760bde16b79ee9bc9263a99a7830961f845f516"}]},"branch":"refs/heads/master"},"75c7d75514bf78474e781082ba8c468fc9653a3c":{"kind":"REWORK","_number":3,"created":"2021-09-03 09:11:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/806593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/806593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/806593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/806593/3"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:04:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-03 09:11:44.000000000","tz":120},"subject":"Release final python-tackerclient for Xena","message":"Release final python-tackerclient for Xena\n\nThis proposed a release for python-tackerclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 1.7.0..98789f8\n98789f8 Multi version API support\n6325127 Remove use of testtools.helpers.safe_hasattr\ne8280f5 Remove broken link from Python-TackerClient guide\n94e28fc Correct YAML load warning\nc20ef24 vnfConfigurableProperties missing in VNF show\n4c59029 Update min version of tox\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I82ebacef170592867182de6ecf7e8176447cdaa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75c7d75514bf78474e781082ba8c468fc9653a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75c7d75514bf78474e781082ba8c468fc9653a3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
