)]}'
{"id":"openstack%2Freleases~806594","triplet_id":"openstack%2Freleases~master~Ic98587141f737b65a2bd57be50b08a1c8b6bdd44","project":"openstack/releases","branch":"master","topic":"xena-milestone-3","hashtags":[],"change_id":"Ic98587141f737b65a2bd57be50b08a1c8b6bdd44","subject":"Release final python-troveclient for Xena","status":"MERGED","created":"2021-08-30 14:05:36.000000000","updated":"2021-09-01 13:22:12.000000000","submitted":"2021-09-01 13:22:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806594-xena-milestone-3","meta_rev_id":"6c523600d4474ff0134dc5edb590632ee5a6eca2","_number":806594,"virtual_id_number":806594,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 13:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-01 12:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-09-01 10:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-01 11:10:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-09-01 12:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-01 11:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 14:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 11:20:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-08-31 12:38:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-08-31 14:52:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-09-01 12:51:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"42fb02301785035d6e3fd9b98d8e340f8d0b5c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 14:05:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75c350bc072a4ca4c781fe247709994a8ff2c2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 14:15:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1a7503c209420f8891b2ca81e4f5db : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73a58675bba64619ad0bbed00927d63b : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cc63ab0d798141faa7a064988cba469a : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"31ec8270faa23980db6436130da332fb425b0330","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-08-31 11:44:51.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"b43e519047176fc87f2e19f1fc17e3be22591e50","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 11:46:19.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"4f1f91537b6e67c6d97cab21041c8cd995bdd1d2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-31 12:38:21.000000000","message":"Patch Set 1: Code-Review+2\n\nFor me the listed two change suggest only PATCH version bump, but i don\u0027t have strong opinion in this one. Plus the patch is approved by PTL and no issues found in test job logs. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"6afb4db5036c17a213f9e961736ed3a417fc48c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-31 12:50:31.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e For me the listed two change suggest only PATCH version bump, but i don\u0027t have strong opinion in this one. Plus the patch is approved by PTL and no issues found in test job logs. LGTM.\n\nI wasn\u0027t sure about how to reflect the microversion change.","accounts_in_message":[],"_revision_number":1},{"id":"dfd7ce926157eaaf6a0793c50e1f500c38199015","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-31 14:52:20.000000000","message":"Patch Set 1:\n\nThe tox bump is a development detail. So this only contains one bug fix. I think a 7.1.1 would make more sense than 7.2.0.","accounts_in_message":[],"_revision_number":1},{"id":"bb3a5e91960116d867bf43825b0769b481507a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 08:13:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1541ef5d2f5a1db14e91153dd9d3e2b1e670f0c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 08:14:00.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e The tox bump is a development detail. So this only contains one bug fix. I think a 7.1.1 would make more sense than 7.2.0.\n\nYes you are right. PS2 propose a 7.1.1","accounts_in_message":[],"_revision_number":2},{"id":"414694592d4b6fea5f6c99f9c20ee81fff377304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 08:29:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50a27c2962e45568ce6e3bfe8819a17 : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/688c77ddcf2e41979c85c148a7ee16ca : SUCCESS in 7m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a8ea5b4d92d94d8587a239b2d05a127c : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e41c9b1a82b3bcec070e111bd1412f4980821473","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-01 10:48:26.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"a01d050a24898cfae46a6c8112b89a52016d27f4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-09-01 11:10:44.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks again for taking care of that!","accounts_in_message":[],"_revision_number":2},{"id":"1a2b1c1b6cc93613b668f04040a087ab98813d06","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:12:01.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"e8e282ea4987d7ba250ed1f7bbb01bda95a7e366","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-01 12:51:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4c0ffe3f2a3006010fb06a496bddb7767a371957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 12:51:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fc1026d18e16ebb2a82e5998b1ed9b646498cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 13:22:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deeaaceca936468a88815bb69b8ba2ee : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc207eca4154ac9a591aa88ddbf0ae2 : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c15d91cb0054248b7dba76df86a43e4 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6c523600d4474ff0134dc5edb590632ee5a6eca2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 13:22:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e80f5a7678e356f757ce83fc79f0991bca17c60","revisions":{"7200fe8f52ba30bd54e74fa0abf85a4d94cd9557":{"kind":"REWORK","_number":1,"created":"2021-08-30 14:05:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/94/806594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/806594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/806594/1"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:05:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:05:25.000000000","tz":120},"subject":"Release final python-troveclient for Xena","message":"Release final python-troveclient for Xena\n\nThis proposed a release for python-troveclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 7.1.0..64f3f62\n64f3f62 Fix detach configuration from instance\n5ce2e99 Changed minversion in tox to 3.18.0\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ic98587141f737b65a2bd57be50b08a1c8b6bdd44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7200fe8f52ba30bd54e74fa0abf85a4d94cd9557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7200fe8f52ba30bd54e74fa0abf85a4d94cd9557"}]},"branch":"refs/heads/master"},"6e80f5a7678e356f757ce83fc79f0991bca17c60":{"kind":"REWORK","_number":2,"created":"2021-09-01 08:13:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/94/806594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/806594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/806594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/806594/2"}}},"commit":{"parents":[{"commit":"89961fff6816c9d5521e4855112b96a84f5f0181","subject":"Merge \"[manila][yoga] Add project tracking schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89961fff6816c9d5521e4855112b96a84f5f0181"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-08-30 14:05:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-01 08:13:16.000000000","tz":120},"subject":"Release final python-troveclient for Xena","message":"Release final python-troveclient for Xena\n\nThis proposed a release for python-troveclient to get any final merged changes\nout prior to the lib freeze for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 02.\n\n$ git log --oneline --no-merges 7.1.0..64f3f62\n64f3f62 Fix detach configuration from instance\n5ce2e99 Changed minversion in tox to 3.18.0\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ic98587141f737b65a2bd57be50b08a1c8b6bdd44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e80f5a7678e356f757ce83fc79f0991bca17c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e80f5a7678e356f757ce83fc79f0991bca17c60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
