)]}'
{"id":"openstack%2Freleases~807048","triplet_id":"openstack%2Freleases~master~If5950c9cc02d7aaf4bf66bc70626d0026bb7acbc","project":"openstack/releases","branch":"master","topic":"oslo-indep","hashtags":[],"change_id":"If5950c9cc02d7aaf4bf66bc70626d0026bb7acbc","subject":"[oslo independent] release etcd3gw","status":"ABANDONED","created":"2021-09-02 08:47:19.000000000","updated":"2021-09-16 13:47:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08b5a8b94165995baf4ad4cf4b3834cd7cd81727","_number":807048,"virtual_id_number":807048,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-02 09:00:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-02 08:48:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 08:48:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 19:53:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"}],"messages":[{"id":"1a092fa6783d59db0adbfa3e53408809904fb6e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 08:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d30a6c8ea5a12ec806091adb507e704bb9755a25","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 08:48:44.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2bd58d258aaae222f42eb4f2fe06b5935e6cbef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 09:00:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182792eb46e54236a4069d639eedd8fe : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6413e2ffd9b540f0b9808d9ebe77b4e5 : SUCCESS in 5m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/87917710a66f43be9661fa2b28087675 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a0c9c9dad74ba7131abf9ce46fe5b6adff30bcd8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-03 19:53:14.000000000","message":"Patch Set 1:\n\nLooks fine, but I\u0027m not sure we really need to release this right now. There\u0027s only one small change to remove explicit unicode strings in doc and release note configs: https://review.opendev.org/c/openstack/etcd3gw/+/768982\n\nI\u0027d say it\u0027s probably better to just hold off and release this once there\u0027s actually something interesting to include.","accounts_in_message":[],"_revision_number":1},{"id":"08b5a8b94165995baf4ad4cf4b3834cd7cd81727","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-16 13:47:19.000000000","message":"Abandoned\n\nNo significant changes landed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d267f64873c6042cd0352dcb9c8dac003e83ab7","revisions":{"1d267f64873c6042cd0352dcb9c8dac003e83ab7":{"kind":"REWORK","_number":1,"created":"2021-09-02 08:47:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/48/807048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/807048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/807048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/807048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/807048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/807048/1"}}},"commit":{"parents":[{"commit":"5afcba66cabc6f3841721ddec176ad7ef439e6b6","subject":"Merge \"Add Glance Xena highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5afcba66cabc6f3841721ddec176ad7ef439e6b6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-02 08:47:09.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-02 08:47:09.000000000","tz":120},"subject":"[oslo independent] release etcd3gw","message":"[oslo independent] release etcd3gw\n\nChange-Id: If5950c9cc02d7aaf4bf66bc70626d0026bb7acbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d267f64873c6042cd0352dcb9c8dac003e83ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d267f64873c6042cd0352dcb9c8dac003e83ab7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
