)]}'
{"id":"openstack%2Freleases~807391","triplet_id":"openstack%2Freleases~master~I42f76ad35c166b70a8ce1b34828a2f4f0d96ffa9","project":"openstack/releases","branch":"master","topic":"cycle-highlights","hashtags":[],"change_id":"I42f76ad35c166b70a8ce1b34828a2f4f0d96ffa9","subject":"[manila] Add Xena cycle highlights","status":"MERGED","created":"2021-09-03 18:32:40.000000000","updated":"2021-09-10 10:42:32.000000000","submitted":"2021-09-10 10:42:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807391-cycle-highlights","meta_rev_id":"5f5fede9ae00dc1b4207602855a3c3a38713d1db","_number":807391,"virtual_id_number":807391,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 10:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-09-10 10:29:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-10 08:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-09 20:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-09-10 10:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-09 19:29:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":32070,"name":"Helena Spease","email":"helena@openstack.org","username":"hspease"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 18:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 19:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-09-04 22:31:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-09-07 17:13:21.000000000","updated_by":{"_account_id":32070,"name":"Helena Spease","email":"helena@openstack.org","username":"hspease"},"reviewer":{"_account_id":32070,"name":"Helena Spease","email":"helena@openstack.org","username":"hspease"},"state":"CC"},{"updated":"2021-09-10 08:07:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-10 10:29:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7980aef32ad677ce0b72952c1371e8ad32462a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 18:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1faf6c190fea80994a1084164777c64f0214a3c8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-03 18:33:19.000000000","message":"Patch Set 1: Workflow-1\n\nWe\u0027re still getting some of these in - so marking this WIP, review, but don\u0027t merge please :)","accounts_in_message":[],"_revision_number":1},{"id":"451ceb7aef05a4eb2e5fe7c5a6fbc67d279e6c9d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:33:48.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2b0498c6e2f81c9d54460d2e3b1f5e8bbc218496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:43:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd7c4360b024226b90ceef6c422fc15 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/365851abd2b64b52ad376da778852fcf : SUCCESS in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/57bee49a0a2045d291b7c204e3c764f1 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9cd3029f0a10dc6f2aeabb1ed5eb3cda15053cf2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-04 22:31:53.000000000","message":"Patch Set 1:\n\nMight be good to just do what is already merged, then follow up later if you get more done. That way the marketing engine has some input at least.","accounts_in_message":[],"_revision_number":1},{"id":"d6e07554cbb29123c6f87b804dc26b5d0ee0f360","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-07 17:05:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Might be good to just do what is already merged, then follow up later if you get more done. That way the marketing engine has some input at least.\n\nHi Sean, \n\nHelena is aware of this patch :)","accounts_in_message":[],"_revision_number":1},{"id":"c7ad36bf1c59b67db05a81eba59e09a6975cf413","author":{"_account_id":32070,"name":"Helena Spease","email":"helena@openstack.org","username":"hspease"},"date":"2021-09-07 17:13:21.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Might be good to just do what is already merged, then follow up later if you get more done. That way the marketing engine has some input at least.\n\u003e \n\u003e Hi Sean, \n\u003e \n\u003e Helena is aware of this patch :)\n\nYep! I am keeping my eye on it","accounts_in_message":[],"_revision_number":1},{"id":"cf0f270f4bf0081a986ef853aca62043cbc99218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-09 19:28:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f41f0af559aac4a57332ba76a8d39441ee17039","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 19:29:31.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"96e1b33c86055432b0bb733654c15a06301cb34e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-09 19:37:19.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Might be good to just do what is already merged, then follow up later if you get more done. That way the marketing engine has some input at least.\n\u003e \u003e \n\u003e \u003e Hi Sean, \n\u003e \u003e \n\u003e \u003e Helena is aware of this patch :)\n\u003e \n\u003e Yep! I am keeping my eye on it\n\nSorry for the wait - it was worth keeping this one open because we deferred one of the features to the Yoga release :) I\u0027ve updated the highlights now, this is good to go.\n\nThanks Helena, Sean!","accounts_in_message":[],"_revision_number":2},{"id":"6aa57d6d4bc8087e2cedd31bcd2c89daf2162923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 19:38:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14395260d5364c2fbfa285f3c0267655 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20dd12570ab343d2bd7c880f1aac38ab : SUCCESS in 6m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/17bf5ba17e0a4a7f8012eb97304fd5c3 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e96e5d7ac4b96d489926ca2ff51c9da933a742fb","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-09-09 20:48:21.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"129b5ac5740066a0beb87d3b121310f5363b7a6e","author":{"_account_id":32070,"name":"Helena Spease","email":"helena@openstack.org","username":"hspease"},"date":"2021-09-09 20:51:35.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Might be good to just do what is already merged, then follow up later if you get more done. That way the marketing engine has some input at least.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hi Sean, \n\u003e \u003e \u003e \n\u003e \u003e \u003e Helena is aware of this patch :)\n\u003e \u003e \n\u003e \u003e Yep! I am keeping my eye on it\n\u003e \n\u003e Sorry for the wait - it was worth keeping this one open because we deferred one of the features to the Yoga release :) I\u0027ve updated the highlights now, this is good to go.\n\u003e \n\u003e Thanks Helena, Sean!\n\nNo worries! Thank you for all your work on this :)","accounts_in_message":[],"_revision_number":2},{"id":"4a4d748cef2700933cb4a6fd5b92b305ac52dde5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 08:07:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"493435bdb66fd5fb1d6e4d5433c3f1f481da86ac","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-10 10:29:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"9e326a761d8bd14284544a2f652599b602cc333d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:29:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b9d862b77dd082534376cae46969cafdbab82ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:42:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b25fc0f909c4d0ba2f03160a1eb029f : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff9995e5366443194b67d3da4794334 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b98710c3ce54dffbdd37b97927f8b70 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5f5fede9ae00dc1b4207602855a3c3a38713d1db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:42:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d4d94b60308ae3bcdbe36aaff5eb0564fbd5b47","revisions":{"101b35dd4aed7a1cee6d3c05c4e4796d955fa28c":{"kind":"REWORK","_number":1,"created":"2021-09-03 18:32:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/91/807391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/807391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/807391/1"}}},"commit":{"parents":[{"commit":"1e8b1d7291b7939cf9cb3745f0c7f9c4a9ae929b","subject":"Merge \"Release final python-cinderclient for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e8b1d7291b7939cf9cb3745f0c7f9c4a9ae929b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-03 18:32:23.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-03 18:32:23.000000000","tz":-420},"subject":"[manila] Add Xena cycle highlights","message":"[manila] Add Xena cycle highlights\n\nChange-Id: I42f76ad35c166b70a8ce1b34828a2f4f0d96ffa9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/101b35dd4aed7a1cee6d3c05c4e4796d955fa28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/101b35dd4aed7a1cee6d3c05c4e4796d955fa28c"}]},"branch":"refs/heads/master"},"9d4d94b60308ae3bcdbe36aaff5eb0564fbd5b47":{"kind":"REWORK","_number":2,"created":"2021-09-09 19:28:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/91/807391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/807391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/807391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/807391/2"}}},"commit":{"parents":[{"commit":"1e8b1d7291b7939cf9cb3745f0c7f9c4a9ae929b","subject":"Merge \"Release final python-cinderclient for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e8b1d7291b7939cf9cb3745f0c7f9c4a9ae929b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-03 18:32:23.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-09-09 19:27:52.000000000","tz":-420},"subject":"[manila] Add Xena cycle highlights","message":"[manila] Add Xena cycle highlights\n\nChange-Id: I42f76ad35c166b70a8ce1b34828a2f4f0d96ffa9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d4d94b60308ae3bcdbe36aaff5eb0564fbd5b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d4d94b60308ae3bcdbe36aaff5eb0564fbd5b47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
