)]}'
{"id":"openstack%2Freleases~807627","triplet_id":"openstack%2Freleases~master~Ia54ccf4f329525e7cd9fe739ccd7bf04969d90d0","project":"openstack/releases","branch":"master","topic":"xena-stable-branches","hashtags":[],"change_id":"Ia54ccf4f329525e7cd9fe739ccd7bf04969d90d0","subject":"[manila] Create for all client and non-client lib xena branch","status":"MERGED","created":"2021-09-06 20:17:59.000000000","updated":"2021-09-10 14:23:34.000000000","submitted":"2021-09-10 14:23:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807627-xena-stable-branches","meta_rev_id":"f5c9983af0a859e1a1d6b4c58dc260e51911d948","_number":807627,"virtual_id_number":807627,"owner":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 14:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-09-09 20:03:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-09-10 10:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-10 07:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-09 20:49:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-10 14:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-09 20:05:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 20:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-06 20:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-09-06 20:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-09-06 20:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 11:40:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-09-10 07:37:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-10 10:13:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"245bf1e83cfe7bc3e3b4bf88c5b9d4fa9e42942d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2021-09-06 20:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8474649c106b5fbf50f5bbb7815da58530dd9403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 20:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1197be2ba0b846049e84e2bab59fdc3e : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a00a2f0b270d4f4191b09e84b15696ac : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/473babf2f6cb49ed811f398c0ebdb720 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bce5962e9525a7fcab9c470cae5537f84cc4ebf5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-09 11:40:39.000000000","message":"Patch Set 1:\n\nHello manila team,\n\nDeadline is today, so please can you check this patch and raise your hands if something on your side is needed.\n\nThanks by advance.","accounts_in_message":[],"_revision_number":1},{"id":"bf6ce6ad8ec82931db3ba565c80119baa4e4c8da","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-09 19:31:08.000000000","message":"Patch Set 1:\n\nHi Herve, \n\nwe have merged some more patches into python-manilaclient \nDo we tag a new release and set that as stable/xena?","accounts_in_message":[],"_revision_number":1},{"id":"0658ec5938e0ab77682b44353838142af7909be0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-09 19:34:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Hi Herve, \n\u003e \n\u003e we have merged some more patches into python-manilaclient \n\u003e Do we tag a new release and set that as stable/xena?\n\nJust one commit, but that adds a new option to an existing command:\n\ngit log --oneline 3.0.0.. --no-merges\n9a2bcb8d8 Add --scheduler_hints to share create command","accounts_in_message":[],"_revision_number":1},{"id":"f2d8da66306f78fef989b7a7ffc84ab603b7232e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-09 20:03:49.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Hi Herve, \n\u003e \u003e \n\u003e \u003e we have merged some more patches into python-manilaclient \n\u003e \u003e Do we tag a new release and set that as stable/xena?\n\u003e \n\u003e Just one commit, but that adds a new option to an existing command:\n\u003e \n\u003e git log --oneline 3.0.0.. --no-merges\n\u003e 9a2bcb8d8 Add --scheduler_hints to share create command\n\n\nFollow up: I asked on #openstack-releases about tagging another release by picking up this commit. The commit message here made me assume that was trivially possible :) However, smcginnis and fungi help me understand the process better, and pointed me to the call out from Hervé last week:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-September/024647.html\n\n\nWhile it is useful for xena CLI users to be able to use the new API parameter introduced after the lib freeze, i agree we have deadlines for a reason, and causing a requirements churn to pick this up may not be in the best interest of the OpenStack projects and projects elsewhere that are testing with the latest tag during this feature freeze period. So i\u0027m good to tag this release.","accounts_in_message":[],"_revision_number":1},{"id":"c83dcd23f912c2ce1a8fe52fc55871de51d63b62","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 20:05:07.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"90b7f4e5af9bd5041354fc42a34df4ae1794f97d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-09-09 20:49:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f80db9a7f27db7932d3bd1bfc3a92dd92b1f8c15","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 07:37:03.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hi Herve, \n\u003e \u003e \u003e \n\u003e \u003e \u003e we have merged some more patches into python-manilaclient \n\u003e \u003e \u003e Do we tag a new release and set that as stable/xena?\n\u003e \u003e \n\u003e \u003e Just one commit, but that adds a new option to an existing command:\n\u003e \u003e \n\u003e \u003e git log --oneline 3.0.0.. --no-merges\n\u003e \u003e 9a2bcb8d8 Add --scheduler_hints to share create command\n\u003e \n\u003e \n\u003e Follow up: I asked on #openstack-releases about tagging another release by picking up this commit. The commit message here made me assume that was trivially possible :) However, smcginnis and fungi help me understand the process better, and pointed me to the call out from Hervé last week:\n\u003e \n\u003e http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024647.html\n\u003e \n\u003e \n\u003e While it is useful for xena CLI users to be able to use the new API parameter introduced after the lib freeze, i agree we have deadlines for a reason, and causing a requirements churn to pick this up may not be in the best interest of the OpenStack projects and projects elsewhere that are testing with the latest tag during this feature freeze period. So i\u0027m good to tag this release.\n\nFirst sorry for my late response.\n\nYes we past feature freeze and the requirement freeze and these changes (9a2bcb8d8) are a new feature so, since we are close to the deadline and for the sack of stability I would suggest to let pass them and let\u0027s cut xena. \n\nThanks for your previous feedback.","accounts_in_message":[],"_revision_number":1},{"id":"7c8d03ec2556fbc34161e1243b91f473a0cf924f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-10 10:13:51.000000000","message":"Patch Set 1: Code-Review+2\n\nThe branch cut looks OK to me, PTL approved the patch.","accounts_in_message":[],"_revision_number":1},{"id":"2e75150695e779cdbba5d475ebf9f7542f5c0901","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 14:06:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"117d8190904c1524044f3ed453c04d381ed3b1fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca1e1981ceae534f14c46e201830ba8a8dc5c3cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:23:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1538f00e98d94402a29c72c97c226207 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a24b7d8a8e0482ebd8be14b89e0a670 : SUCCESS in 8m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2993b57a8b16468e8805080dd2892974 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f5c9983af0a859e1a1d6b4c58dc260e51911d948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:23:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ef1b1bba5cb0eeb1d1999f0a4f6ac6fc0acfa36","revisions":{"0ef1b1bba5cb0eeb1d1999f0a4f6ac6fc0acfa36":{"kind":"REWORK","_number":1,"created":"2021-09-06 20:17:59.000000000","uploader":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"ref":"refs/changes/27/807627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/807627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/807627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/807627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/807627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/807627/1"}}},"commit":{"parents":[{"commit":"8e2748c042c8b6f9aa06507a1363ddec6c16ed6a","subject":"Relase osc-placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e2748c042c8b6f9aa06507a1363ddec6c16ed6a"}]}],"author":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-09-06 20:17:20.000000000","tz":-240},"committer":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-09-06 20:17:20.000000000","tz":-240},"subject":"[manila] Create for all client and non-client lib xena branch","message":"[manila] Create for all client and non-client lib xena branch\n\nThis creates the stable/xena branch for all client\nand non-client libraries now that we are past the\nlib freeze on September 02.\n\nIf a team wants to wait for a specific patch to make it to the library,\nsomeone from the team can -1 the patch to have it held,or update that\npatch with a different commit SHA.\n\nPatches with no response will be merged by Friday, September 10.\n\nChange-Id: Ia54ccf4f329525e7cd9fe739ccd7bf04969d90d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ef1b1bba5cb0eeb1d1999f0a4f6ac6fc0acfa36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ef1b1bba5cb0eeb1d1999f0a4f6ac6fc0acfa36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
