)]}'
{"id":"openstack%2Freleases~808494","triplet_id":"openstack%2Freleases~master~I343a4153dcaed01ef6f827a894c21a053c3cc5eb","project":"openstack/releases","branch":"master","topic":"osh-xena","hashtags":[],"change_id":"I343a4153dcaed01ef6f827a894c21a053c3cc5eb","subject":"[openstack-helm] Add Xena cycle highlights","status":"ABANDONED","created":"2021-09-10 20:44:36.000000000","updated":"2021-09-13 14:00:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2d8c2b828028ba9cccb54cc596d2b8918e318b15","_number":808494,"virtual_id_number":808494,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-10 21:28:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2021-09-13 09:04:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2021-09-13 07:27:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 20:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 07:27:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-13 09:04:07.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"54f039434b2114fd3e376d2e7d0876633a10efeb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-10 20:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fabb5953d44fdce2d3b8061aa841d52cd25d4d7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 20:46:07.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"4d9af2d54f249af3d5181b1039270ef34462821d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-10 20:47:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"570a6848ff424aa01e43d3b74093086f2323f184","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-10 20:47:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a2a33f01b457b5e0b713efa135c03ba5ce1e6d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 20:58:12.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8798b434b54e2094c4f02ab425f3f9 : SUCCESS in 9m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df2516b93a29417ab117e718139c842b : FAILURE in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/891548f8197c4b0ea3dcf0dea60879ff : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8690da53e7d616e27cc3843aa95c63bf93bdfb38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-10 20:59:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c7aebbca79487f5e9b6824b83b31e05e7a3b2a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:17:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d61736d215a49aca1771fe7e4846a1b : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6a9b06feb354513a67c8f57a73fdf9a : FAILURE in 6m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d0f2bd1e8e484b9287127549f80e0b0a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"59bf7786ac7725d35a3991fae750b4e8c99aebf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-10 21:19:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"689d2fe524122f6e0ebfbe7c26dd6d1e85810577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:28:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d1eaeab45845dfbf1027889272c745 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea73423de9f141e29cb0f571c0b73dd4 : SUCCESS in 6m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c99141b895104760b5db0fc8a788bdc7 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":4},{"id":"d3bd26eaefaf2a99735c57312e2cc2ee834e3846","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 07:27:10.000000000","message":"Patch Set 4: Code-Review-1\n\nHello,\n\nopenstack-help is not managed by the coordinated release.\n\nhttps://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L1954","accounts_in_message":[],"_revision_number":4},{"id":"406a326a2818c0c085fde5edb5952b59cfe98c1f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 07:27:34.000000000","message":"Patch Set 4:\n\ns/openstack-help/openstack-helm/","accounts_in_message":[],"_revision_number":4},{"id":"a07880d4735b3d32508772c91a3369b421510402","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-13 09:04:07.000000000","message":"Patch Set 4: Code-Review-1\n\nWe could try to find a way to support release-highlights that are not releaseteam-managed but for now I agree, this won\u0027t work","accounts_in_message":[],"_revision_number":4},{"id":"5737aa127e058e1f888a94b2ae908e5bead33fc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 10:30:10.000000000","message":"Patch Set 4:\n\nFrom a Xena viewpoint it\u0027s a bit late to introduce this kind of support.\nIt could be done by adding a new deliverable file schema on our side to inform our tooling to skip our classical checks. However, another point to address is how to reflect that without misleading people in the governance informations. Maybe by introducing another and separated repository?","accounts_in_message":[],"_revision_number":4},{"id":"2d8c2b828028ba9cccb54cc596d2b8918e318b15","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-13 14:00:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"598a4870b90d166d1eb762159f5caf8b57f148e5","revisions":{"476751f41ef9f4824cc5912231c07b2a734f6860":{"kind":"REWORK","_number":1,"created":"2021-09-10 20:44:36.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/94/808494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/808494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/808494/1"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:44:36.000000000","tz":0},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:44:36.000000000","tz":0},"subject":"[openstack-helm] Add Xena cycle highlights","message":"[openstack-helm] Add Xena cycle highlights\n\nChange-Id: I343a4153dcaed01ef6f827a894c21a053c3cc5eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/476751f41ef9f4824cc5912231c07b2a734f6860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/476751f41ef9f4824cc5912231c07b2a734f6860"}]},"branch":"refs/heads/master"},"cb21e818e3b51819b1b848418cb9c5c6ad6f4d7f":{"kind":"REWORK","_number":2,"created":"2021-09-10 20:47:47.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/94/808494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/808494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/808494/2"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:44:36.000000000","tz":0},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:47:41.000000000","tz":0},"subject":"[openstack-helm] Add Xena cycle highlights","message":"[openstack-helm] Add Xena cycle highlights\n\nChange-Id: I343a4153dcaed01ef6f827a894c21a053c3cc5eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb21e818e3b51819b1b848418cb9c5c6ad6f4d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb21e818e3b51819b1b848418cb9c5c6ad6f4d7f"}]},"branch":"refs/heads/master"},"b799156eb18951caa9c8517d31eef673c78f3cad":{"kind":"REWORK","_number":3,"created":"2021-09-10 20:59:21.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/94/808494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/808494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/808494/3"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:44:36.000000000","tz":0},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:59:20.000000000","tz":0},"subject":"[openstack-helm] Add Xena cycle highlights","message":"[openstack-helm] Add Xena cycle highlights\n\nChange-Id: I343a4153dcaed01ef6f827a894c21a053c3cc5eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b799156eb18951caa9c8517d31eef673c78f3cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b799156eb18951caa9c8517d31eef673c78f3cad"}]},"branch":"refs/heads/master"},"598a4870b90d166d1eb762159f5caf8b57f148e5":{"kind":"REWORK","_number":4,"created":"2021-09-10 21:19:13.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/94/808494/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/808494/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/808494/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/808494/4"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 20:44:36.000000000","tz":0},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-09-10 21:19:12.000000000","tz":0},"subject":"[openstack-helm] Add Xena cycle highlights","message":"[openstack-helm] Add Xena cycle highlights\n\nChange-Id: I343a4153dcaed01ef6f827a894c21a053c3cc5eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/598a4870b90d166d1eb762159f5caf8b57f148e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/598a4870b90d166d1eb762159f5caf8b57f148e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
