)]}'
{"id":"openstack%2Freleases~808600","triplet_id":"openstack%2Freleases~master~I16f87d6442dc21c308c3241806f19a5f1d4e6398","project":"openstack/releases","branch":"master","topic":"xena-rc1-deadline","hashtags":[],"change_id":"I16f87d6442dc21c308c3241806f19a5f1d4e6398","subject":"Release Xena RC1 for barbican","status":"MERGED","created":"2021-09-13 08:04:45.000000000","updated":"2021-09-16 10:54:19.000000000","submitted":"2021-09-16 10:54:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808600-xena-rc1-deadline","meta_rev_id":"5f5f544d80c8fc818d81fec0c5f9f4239c15d5d4","_number":808600,"virtual_id_number":808600,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 10:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-09-16 10:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-16 07:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-09-16 10:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-16 10:54:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 08:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-13 08:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 11:24:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-09-13 11:24:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11561,"name":"Dave McCowan","email":"dmccowan@cisco.com","username":"dave-mccowan"},"state":"REVIEWER"},{"updated":"2021-09-16 10:40:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"95b8a84947aa1d1b6b8263abf07f1534cd1c3361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 08:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5709a535c080bd9f5b9919e0561e48fe97d1aadd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:19:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef9c13ee7834ca1a5d56f4f3b0a4a63 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f4cbb4033d2475e85b8da94dbc89f42 : SUCCESS in 7m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1043bde26ba24764b85a7d6aea5ce359 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"980c11119a005999f8f687b8975522bb81fa136c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-13 16:45:22.000000000","message":"Patch Set 1: Code-Review-1\n\nI would like to wait for this bugfix to merge before releasing RC1: https://review.opendev.org/c/openstack/barbican/+/796065","accounts_in_message":[],"_revision_number":1},{"id":"691002f804580b255fb5fb3a6bd86d953869bb70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-15 18:45:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca07c42db263c8cca12b43e51bdfd855701cb8af","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 18:46:41.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8d42fe9b5a93cc06e1eeddf318ce5db0e9a1dbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 18:56:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6b9115eee546279741f6be61efd290 : SUCCESS in 9m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31a6c104e99c4c85816873091b9f5d2a : SUCCESS in 8m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/413571997b34411689cdb81574ca74f2 : FAILURE in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4ce4d2a9c25a1b1a371264985c2d1e358007c593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-09-15 19:04:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"89b375da9e311bc566b6a407e9c9d6fefe55346a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 19:17:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54877dd48dfe4fe0a16a3cd0ab85802b : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffbaa738df044f1482d7edb6e788dc7b : SUCCESS in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6897de3d8fd243c683cda74fcc9993f5 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"738be6157d646d2c0e50bf39e106125189aab0a1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-16 07:54:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e424a5e1403f962d43f629b472cffb3529b3c5b0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-16 10:40:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM:\n* MAJOR version bump is needed\n* patch was updated by PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"191a3a943b95a73bb883060db76acac8c57b4255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:41:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0794204fcc042d284b753353ec4f9245105f2e1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:54:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b743cea646b4906952c6208d90cf9a5 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade60929dd3d4ea0881bc573482519de : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a45046cbd0604e5bb7aad207dc715881 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fab6c1c5dd65c558e32217fffc7e09d2d9e569a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:54:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8762e3911fd43f6ecaae8a7d3784bdc6907c19c3","revisions":{"63bb6cd5d480c4a20dcf4ca38c847e619ecd06ff":{"kind":"REWORK","_number":1,"created":"2021-09-13 08:04:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/00/808600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/808600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/808600/1"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 08:04:38.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 08:04:38.000000000","tz":120},"subject":"Proposing Xena RC1 for barbican","message":"Proposing Xena RC1 for barbican\n\nThis proposed a RC1 release for barbican to get any merged changes\nout prior to the RC1 deadline for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 16.\n\n$ git log --oneline --no-merges 12.0.0.0rc1..200bff89\n200bff89 Return 403 instead of 500 when policy check fails\naa7bf886 Replace oslo_utils.fnmatch with fnmatch\n3a76ced8 Fix alembic migrations\n1ae4badb Fix unit tests and migration to unblock gate\n94a45c1e Add FIPS gate job\nb5d10b3c docs: Update Freenode to OFTC\nc59f2a6b Raise maximum allowed secret size\n6d0df114 Imported Translations from Zanata\n6662fc9e setup.cfg: Replace dashes with underscores\n21661beb Fix Vault functional test\nced3e5c0 Add secure-rbac gate\n922c68ba Fix RBAC and ACL access for managing secret containers\n57f334e0 Fix transport key policies\n2e8d7dec Add Python3 xena unit tests\na567702c Update master for stable/wallaby\n672dc0b5 Add ACL default to allow project read\n14e3ab97 Monkey patch original current_thread _active\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I16f87d6442dc21c308c3241806f19a5f1d4e6398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63bb6cd5d480c4a20dcf4ca38c847e619ecd06ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63bb6cd5d480c4a20dcf4ca38c847e619ecd06ff"}]},"branch":"refs/heads/master"},"b7e664d1faec5b448494254e38bae48149dbb291":{"kind":"REWORK","_number":2,"created":"2021-09-15 18:45:19.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/00/808600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/808600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/808600/2"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 08:04:38.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-15 18:42:25.000000000","tz":-300},"subject":"Release Xena RC1 for barbican","message":"Release Xena RC1 for barbican\n\n$ git log --oneline --no-merges 12.0.0.0rc1..b9daa100\nb9daa100 Fix Castellan Secret Store inconsistent encoding\n200bff89 Return 403 instead of 500 when policy check fails\naa7bf886 Replace oslo_utils.fnmatch with fnmatch\n3a76ced8 Fix alembic migrations\n1ae4badb Fix unit tests and migration to unblock gate\n94a45c1e Add FIPS gate job\nb5d10b3c docs: Update Freenode to OFTC\nc59f2a6b Raise maximum allowed secret size\n6d0df114 Imported Translations from Zanata\n6662fc9e setup.cfg: Replace dashes with underscores\n21661beb Fix Vault functional test\nced3e5c0 Add secure-rbac gate\n922c68ba Fix RBAC and ACL access for managing secret containers\n57f334e0 Fix transport key policies\n2e8d7dec Add Python3 xena unit tests\na567702c Update master for stable/wallaby\n672dc0b5 Add ACL default to allow project read\n14e3ab97 Monkey patch original current_thread _active\n\nChange-Id: I16f87d6442dc21c308c3241806f19a5f1d4e6398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7e664d1faec5b448494254e38bae48149dbb291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7e664d1faec5b448494254e38bae48149dbb291"}]},"branch":"refs/heads/master"},"8762e3911fd43f6ecaae8a7d3784bdc6907c19c3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-15 19:04:43.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/00/808600/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/808600/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/808600/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/808600/3"}}},"commit":{"parents":[{"commit":"77976b7e619ec7905fc53eec8977012cf318fc52","subject":"Merge \"Release OpenStack-Ansible Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77976b7e619ec7905fc53eec8977012cf318fc52"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 08:04:38.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-09-15 19:04:43.000000000","tz":0},"subject":"Release Xena RC1 for barbican","message":"Release Xena RC1 for barbican\n\n$ git log --oneline --no-merges 12.0.0.0rc1..b9daa100\nb9daa100 Fix Castellan Secret Store inconsistent encoding\n200bff89 Return 403 instead of 500 when policy check fails\naa7bf886 Replace oslo_utils.fnmatch with fnmatch\n3a76ced8 Fix alembic migrations\n1ae4badb Fix unit tests and migration to unblock gate\n94a45c1e Add FIPS gate job\nb5d10b3c docs: Update Freenode to OFTC\nc59f2a6b Raise maximum allowed secret size\n6d0df114 Imported Translations from Zanata\n6662fc9e setup.cfg: Replace dashes with underscores\n21661beb Fix Vault functional test\nced3e5c0 Add secure-rbac gate\n922c68ba Fix RBAC and ACL access for managing secret containers\n57f334e0 Fix transport key policies\n2e8d7dec Add Python3 xena unit tests\na567702c Update master for stable/wallaby\n672dc0b5 Add ACL default to allow project read\n14e3ab97 Monkey patch original current_thread _active\n\nChange-Id: I16f87d6442dc21c308c3241806f19a5f1d4e6398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8762e3911fd43f6ecaae8a7d3784bdc6907c19c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8762e3911fd43f6ecaae8a7d3784bdc6907c19c3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
