)]}'
{"id":"openstack%2Freleases~808706","triplet_id":"openstack%2Freleases~master~I7aa3b297d766b6e7c05c1cfb4aad56b2fdaea42e","project":"openstack/releases","branch":"master","topic":"xena-rc1-deadline","hashtags":[],"change_id":"I7aa3b297d766b6e7c05c1cfb4aad56b2fdaea42e","subject":"Proposing Xena RC1 for nova","status":"MERGED","created":"2021-09-13 09:30:48.000000000","updated":"2021-09-17 14:19:53.000000000","submitted":"2021-09-17 14:19:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808706-xena-rc1-deadline","meta_rev_id":"62528f7d394a6186f32c72da9e9b317ab9a92f29","_number":808706,"virtual_id_number":808706,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 14:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-17 14:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-17 07:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-16 19:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-09-17 07:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-09-17 14:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-17 07:39:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 09:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-13 09:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 11:21:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-09-14 14:18:01.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-09-17 12:51:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-09-17 14:01:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"531a4b1a2992978b90e6786da4c1ab27b16d9657","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 09:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bef76d665798d9a9bd0fae35fe592c6a4c41c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:49:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b50a06d6624c369b215b120ef15bf0 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/678305c9af3c460ca62d1f423d39acef : SUCCESS in 6m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fdeb8ee428cc443b8b7194ada9431114 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1f9143fbb78e5d01fedd68cd679d1712b5d73e9a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-14 14:18:01.000000000","message":"Patch Set 1: Code-Review-1\n\nwe need the reno prelude patch to land https://review.opendev.org/c/openstack/nova/+/807786","accounts_in_message":[],"_revision_number":1},{"id":"24d12c99fa4cfbe95af5cb2f7440e11098f72473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-16 19:46:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f074ce8a49e589ca70002a81cf71752d4969009f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-16 19:46:51.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e we need the reno prelude patch to land https://review.opendev.org/c/openstack/nova/+/807786\n\nreno prelude landed, I\u0027ve updated the hash and the commit message accordingly. Nova RC1 is good to go now","accounts_in_message":[],"_revision_number":2},{"id":"174712204653e3a7b811825bfb8717a721f16be9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-16 19:46:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"956f04471531aa8648d94e02fb2d9f1fc3cf1734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 20:49:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e51200f573482d9bbbe52b60fb4108 : TIMED_OUT in 31m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/269d752b92b742379950080d8aeab9e6 : TIMED_OUT in 1h 01m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0caff1e90cdc412fa9e37585854fc2de : TIMED_OUT in 1h 01m 18s","accounts_in_message":[],"_revision_number":2},{"id":"42ec974ed91c765558094308aeb2445c3f980fad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-17 05:00:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"40951f564d467c13df4cadb7806471dbedce5dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 06:02:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18353dc7e9c84054b4fa5387793ef3d1 : TIMED_OUT in 31m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b5287f4203647e88e726b9e7f3ad99a : TIMED_OUT in 1h 01m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0fdb4b6b8ce64fac8e6d83963eb0d7ea : TIMED_OUT in 1h 00m 59s","accounts_in_message":[],"_revision_number":2},{"id":"550247bbaf54d73d3c9b712a3b225c537cfde6ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 06:26:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4eb2eb3cef94a8976b4003e7fdb663b14ee637d6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-09-17 07:25:32.000000000","message":"Patch Set 2: Code-Review+1\n\nApproving late...","accounts_in_message":[],"_revision_number":2},{"id":"cd211877123f876d6330e1debb8efccf76fa435e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 07:39:18.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"36370ae31903c7c2cd56391e2211e5f078ced0c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 07:46:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5ad6bbe96233ec8fb9c8d0cd5b717edea9fe54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 07:51:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15f727cfa6a4ef8ab7c6012c0ead6ce : TIMED_OUT in 31m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b832797df7b4ebc9e7bbee3d5ca4f2d : TIMED_OUT in 1h 01m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3b178f59b4e548d398e6c610b129447f : TIMED_OUT in 1h 01m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9baddae9bbc44ef66ca3ce328341e3dac21e4568","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-17 12:51:49.000000000","message":"Patch Set 2:\n\nrecheck - gate is unblocked","accounts_in_message":[],"_revision_number":2},{"id":"2a35d23c62b5b7bd38ac9f43fc750a653ca7e8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:15:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f006e715f64637a768e1c25ffbf73b : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86207363ffcc43e9b489aa887579db09 : SUCCESS in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8a7328215b3d41899ce2bd71da8c6a21 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4d5e8ec61e949600b5bd0b9136446eb2035dd501","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-17 14:01:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* MAJOR version bump is needed\n* approved by PTLs\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":2},{"id":"087467443cb0fd82b326683f6772c5f63245a201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:02:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb599d0db0419dc6141e3ca5425c61d393941fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:19:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d67e8b4e8694fa5815152fbff0daa04 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f5719d8c7f47249d7dedbb82d48058 : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5fe0e2286e304f70bd42c4bfa1b6799c : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":2},{"id":"62528f7d394a6186f32c72da9e9b317ab9a92f29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:19:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19e579e42dcb71974a6566b335de81b5eb22df8f","revisions":{"2c319d967f23df51e20b4c8fbe7b26cc8389352d":{"kind":"REWORK","_number":1,"created":"2021-09-13 09:30:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/06/808706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/808706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/808706/1"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:30:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:30:41.000000000","tz":120},"subject":"Proposing Xena RC1 for nova","message":"Proposing Xena RC1 for nova\n\nThis proposed a RC1 release for nova to get any merged changes\nout prior to the RC1 deadline for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 16.\n\n$ git log --oneline --no-merges 23.0.0.0rc1..a027b45e46\n22d4057a16 [doc] port-resource-request-groups not landed in Xena\nbbcd7d2fd0 fup: Print message logging uncaught nova-manage exceptions\na6ef502aa0 Support Cpu Compararion on Aarch64 Platform\nc590843974 Allow \u0027bochs\u0027 as a display device option\n8541f37cb5 Update compute rpc version alias for xena\nfc77ce191f Add --sleep option for archive_deleted_rows --until-complete\n625469d692 Avoid excessive sleep in vmware unit test\nb67b928c38 Parse alias from domain hostdev\n1b6a6e3916 Convert features not supported error to HTTPBadRequest\n37f80c13c6 [func test] move unshelve test to the proper place\n44309c419f Support interface attach / detach with new resource request format\n122a32ed82 workarounds: Remove rbd_volume_local_attach\n9bd62eae6e workarounds: Remove disable_native_luksv1\n9f8cc2f038 Add two new hacking rules\n9cb92e8832 [func test] refactor interface attach with qos\nfa35cfa998 tests: Address nits for configurable-instance-hostnames series\nc09d98dadb Add force kwarg to delete_allocation_for_instance\n6c5dd864c2 db: Handle parameters in DB strings\ncf07df57e7 Remove module level caching\n191bdf2069 Support move ops with extended resource request\n9be996c696 Expose the mdev class\nc3886c3ca7 Support boot with extended resource request\n2b847085c8 Provide and use other RCs for mdevs if needed\na543cb23aa Provide the mdev class for every PCI device\nfc1a99dc16 docs: admin/networking rename neutron_tunneled to neutron_tunnel\ncfa33d3b06 policy: Deprecate field from \u0027os-extended-server-attributes\u0027 policy\ne906a8c0ec nova-manage: Introduce volume show, refresh, get_connector commands\n191712594d Add some missing parameters in docs of os-cells\n38603a4d09 Reproduce bug 1941005\na057bbc406 fup: Fix os-volume_attachments api-ref parameters\n1d62485472 Fix documentation about cpu topologies\nac21c6674c api: Introduce microversion 2.89 adjusting os-volume_attachments\nb6ab1cb467 api: Log actual number of UUIDs queried\nda6af3fb22 compute: Query the service group API within check_instance_host\n5e8267f703 fup: Increase service_down_time beyond INITIAL_REPORTING_DELAY in test\n50a7a3050a conf: Remove deprecated aliases\n62e112461f tests: Merge \u0027test_utils\u0027, \u0027test_scheduler_utils\u0027\n6652999494 scheduler: Merge driver into manager\n289438b4c2 docs: Drop references to non-filter scheduler drivers\n930b7c9921 Avoid modifying the Mock class in test\n6fbd0b758d address open redirect with 3 forward slashes\n94f47471e0 Transfer RequestLevelParams from ports to scheduling\nd6cd4420bb Parse extended resource request from the port data\n9bae5f2ca5 Add missing [oslo_reports] options\nec6f8f63f9 [func test] ports with both bw and pps resources\nde9b136dc6 [func test] refactor asserts in qos tests\nc9b1dfe9b7 [func test] refactor assertPortMatchesAllocation\nb29ccf8aa5 Support same_subtree in allocation_canadidate query\nf6e8c512fb Bump min placement microversion to 1.36\n9de8bfb6e6 Add same_subtree field to RequestLevelParams\n8f5b7c67ca Reject server operations with extended resource req\ne357ad3c23 Reject server create with extended resource req\n017b0a3d23 Detect port-resource-request-groups neutron API extension\nf68d30bff3 [func test] neutron fixture for extended resource request\naa3d89b31f [func test] move port creation to the NeutronFixture\n713a11727d [func test] create pps resource on OVS agent RP\n1039aa6443 [func test] move port resource request tests\n2383cbb4a5 Fix inactive session error in compute node creation\na8492e8878 Prevent deletion of a compute node belonging to another host\nec48e1523d workarounds: Add libvirt_disable_apic\nb11e3f1d0d fup: Remove unused legacy block_device_info format\n78cace1ce4 fup: Move _migrate_stub into LibvirtMigrationMixin\n8a0ae569ef fixtures: Add legacy APIs to cinder fixture\ncb3e11952d tests: Sanity check some tests\ne3fa6859f9 api: Remove unnecessary \u0027base_\u0027 prefix from schemas\n0942163c82 api: Pass request to \u0027addresses\u0027 view builder\nb543f8226c api: Rename \u0027parameter_types.hostname\u0027 -\u003e \u0027fqdn\u0027\n3f548f3136 VmWare: Fix deprecation warning in unit tests\n1f2aa3e9d8 VMWare: Use get_hardware_devices throughout\n7fc6fe6fae Fix 1vcpu error with multiqueue and vif_type\u003dtap\nd31b26e09d VMWare: Use WithRetrieval to get all results\n39f0af5d18 libvirt: Abort live-migration job when monitoring fails\n20d8554f64 tests: Enable SADeprecationWarning warnings\neb728e877a db: Final cleanups\na7584ec1a5 docs: Add documentation on database migrations\n2226ec4e3b Accelerator smartnic SRIOV support\n2bb4527228 Invalidate provider tree when compute node disappears\n32676a9f45 Clear rebalanced compute nodes from resource tracker\n59d9871e8a Add functional regression test for bug 1853009\n31bedfc09c fup: Assert state of connection_info during LM rollback in func tests\nd4cc10d7bb fup: Make connection_info returned by CinderFixture unique per attachment\nce128798f5 fup: Mock all unix connectors in os-brick fixture\n79ec4e2938 fup: Move OSBrickFixture into base.ServersTestBase\n962eda94d5 compute: Ensure updates to bdms during pre_live_migration are saved\n268b716955 Add a regression test for bug 1939545\nd102b751b7 fake: Ensure need_legacy_block_device_info returns False\n1ba2c1c55d db: Enable auto-generation of migrations\n905c9723e9 db: Integrate alembic\n307adc5e08 db: Normalize migrations tests\nc3e113eb31 db: Trivial style changes\n38a022166f db: Add initial alembic migration for API DB\nca9ba88a7a db: Add initial alembic migration for main DB\n21c5ced0ba db: Avoid use of ALTER in initial migration\n43b253cd60 db: Post reshuffle cleanup\nbf8b5fc7d0 db: Move remaining \u0027nova.db.sqlalchemy\u0027 modules\naec2ca9468 db: Exclude the \u0027use_db_reconnect\u0027 option\n100b9dc62c db: Unify \u0027nova.db.api\u0027, \u0027nova.db.sqlalchemy.api\u0027\n84b6179076 db: Register database config options ourselves\n59c2262ca5 Block servers with vGPU and device profile in heal_allocations\nd4dbcd5fa0 func: Increase rpc_response_timeout in TestMultiCellMigrate tests\n4e18f07670 zuul: Mark live migration jobs as non-voting due to bug #1912310\n9cf07d5aaa Change the admin-visible logs for mdev support\nede7866de4 Add regression test for bug 1938326\nc3245098e3 smartnic support - functional tests\n1f53176d2f smartnic support - reject server move and suspend\ne19fa1a199 smartnic support - cleanup arqs\n120fe8af93 smartnic support - build instance with smartnic arqs\nb90c828d70 smartnic support - create arqs\n8b84135f2d smartnic support - new vnic type\n9c55d5d59d Smartnic support - cyborg drive\n2ffd973860 Add functional test for bug 1937375\n753a2ae5d4 fup: Move _wait_for_volume_attach into InstanceHelperMixin\n2209b0007f compute: Avoid duplicate BDMs during reserve_block_device_name\n49ba5a763f libvirt: Handle silent failures to extend volume within os-brick\nff4d0d002a Rename vgpu options to mdev\ndbf3e276ba zuul: Increase GLANCE_LIMIT_IMAGE_SIZE_TOTAL for nova-lvm\n39ef1ad774 Restore original time override in test_archive_task_logs\n214cabe684 Reduce mocking in test_reject_open_redirect for compat\na3d0aa4772 I2f9ad3df25306e070c8c3538bfed1212d6d8682f fup: add log\n48f35058fe tests: Remove \u0027stub_out_trusted_certs\u0027\nf916266d4f trivial: Cleanup a comment about a now removed libvirt version check\n880611df0b Allow deletion of compute service with no compute nodes\n32257a2a6d Reproducer unit test for bug 1860312\nd4f9b03c5c api: Improve extra spec validator help texts\n1d850a1593 use cinder v3\n809a52a16b Add tests for \u0027AggregateInstanceExtraSpecsFilter\u0027\nf0d17e774f Add tests for \u0027AggregateImagePropertiesIsolation\u0027\n89e321b7b5 tox: Encode specific Python versions\na65ea33824 Bump oslo.db to 10.0.0\n4ee2f667b5 tests: Validate AZ values\n52d9dd2c0b doc: allow_resize_to_same_host needs nova-api restarted\nd093849c3c Correct reno for archiving task_log database records\na22fe2da97 Bump os-resource-classes to 1.1.0\nad60f23be3 extend_volume of libvirt/volume/iscsi should not use device_path\n3bda965e15 tests: Remove unnecessary warnings filter\n0aea2ac903 tests: Silence SQLAlchemy warnings\n1d3373dcf0 Fix request path to query a resource provider by uuid\n982e2ee02d Use neutronclient\u0027s port binding APIs\n87dba379ac Make test_archive_task_logs deterministic\nbe9fe5ce61 Improve prep_resize reschedule unit test\nb84787d26a Add func test for nova-manage db archive_deleted_rows --before\nd88fe2895b docs: Replace \u0027nova boot\u0027 with \u0027openstack server create\u0027\n9cdecc81fb api: Handle invalid volume UUIDs during spawn\nfaa9de93d1 Remove the redundant code for attach interface\n8f21ee42bd api: Align availability zone info with forced host\n9799468d6f db: Drop support for experimental concurrency\n014480d52f db: Remove constant aliases from \u0027nova.db.api\u0027\n089ae234a7 db: Move \u0027sqlalchemy.types\u0027 up a directory\nbcf225daf4 db: Move main DB migrations\n4bbea58cbb db: Move db.sqalchemy.migration to db.migration\n43de2421b3 db: Use module-level imports for sqlalchemy (for real)\n97a566ad9c Add logs when cannot fit numa\ndfda0c0482 Fix oslo policy DeprecatedRule warnings\nf84d5917c6 [ironic] Minimize window for a resource provider to be lost\ne0534cc289 scheduler: Merge \u0027FilterScheduler\u0027 into base class\n6d745036bc trivial: Remove FakeScheduler (for realz)\nc4b4704783 scheduler: Remove \u0027hosts_up\u0027\nb20753b313 scheduler: \u0027USES_ALLOCATION_CANDIDATES\u0027 removal cleanup\n97d25371b3 scheduler: Remove \u0027USES_ALLOCATION_CANDIDATES\u0027\na569a51fed fix sr-iov support on Cavium ThunderX hosts.\n17cca270a6 Make explicit the network backend used in the CI jobs\nfa0fb2fe3d libvirt: Always delegate OVS plug to os-vif\nc31d3f96b1 Fix error \u0027404 Not Found\u0027\n99cf5292c7 objects: Fix VIFMigrateData.supports_os_vif_delegation setter\nb5b2e99824 zuul: Add nova-tox-functional-centos8-py36 job\n085de864db tests: Allow bindep and test-setup.sh to run on EL distros\n387823b36d Fix max cpu topologies with numa affinity\nfe25fa13e0 Test numa and vcpu topologies bug: #1910466\n4de5a92d4e db: Reintroduce validation of shadow table schema\nbc71331bca Add --task-log option to nova-manage db archive_deleted_rows\n98b01c9a59 Move \u0027check-cherry-picks\u0027 test to gate, n-v check\n070e83b749 db: Clean up migration code\nb5dd59f8f9 db: Synchronize function signatures\n1d60cd7e05 db: Copy docs from \u0027nova.db.*\u0027 to \u0027nova.db.sqlalchemy.*\u0027\n7ab2947720 db: Remove \u0027nova.db.base\u0027 module\n5819fe46a9 db: Fold in ForeignKey constraints\n66daa87339 db: Fold in indexes\n3b18b788f2 db: Use module-level imports for sqlalchemy\nd6cad899ca db: Remove unused DB methods\n194df1fa62 db: Remove \u0027nova.db.sqlalchemy.utils\u0027\n6df671fea7 db: Remove dead code\nd966546516 Add test coverage for API version headers in CORS\n6aa580c15f gate: Remove test_evacuate.sh\nb02a95a18b Allow X-OpenStack-Nova-API-Version header in CORS\ne59fc77c3d Retry lvm volume and volume group query\neb67ec24e3 Handle OPERATION_FAILED error during detach\n7c76821499 Add debug log for device detach libvirt error\n31491e5bc3 Fix typos in minimum version policy docs\n7c478ac099 zuul: Skip block migration with attached volumes tests due to bug #1931702\n47bedbb77a Revert \"Removing mypy to fix the nova CI\"\nd71dfa8b8d Move fake_notifier impl under NotificationFixture\n7085770b20 Test the NotificationFixture\n531fa4cd04 Removing mypy to fix the nova CI\n375a8452bb api: Log correct client IP if load balancer in use\n7a0b2d3482 docs: Explicitly state lack of support for node renaming\n5c02091647 requirements: Add types-paramiko\n4d8bf15fec libvirt: Set driver_iommu when attaching virtio devices to SEV instance\n26d65fc882 libvirt: Do not destroy volume secrets during _hard_reboot\nbfde686130 Remove unnecessary mocks from unit test\n7b11409463 Make test_refresh_associations_* deterministic\n0b63f9effe Improve policy doc for supported scope info\n2c761066e0 Remove PROJECT_ADMIN limitation from zero-disk and external-network policy\n1cc52fd2f2 virt: Add destroy_secrets kwarg to destroy and cleanup\n33c8af1f8c Error anti-affinity violation on migrations\n7c7a2a142d Deprecate filters that have been replaced by placement filters\n5cf06bf33d docs: Remove duplicate ToC entry\nfc3838a9b3 Change nova doc after Freenode -\u003e OFTC move\n5c50a45a1e Fix RequestLevelParams persistence handling in RequestSpec\n17ac1a40cb Add test coverage for security group checking\n56eb253e9f Honor [neutron]http_retries in the manual client\ne98994027f hardware: Use image_meta.id within get_mem_encryption_constraint\nd5ed968826 zuul: Skip swap_volume tests as part of nova-next\n740e6f09bf libvirt: Enable \u0027vmcoreinfo\u0027 feature by default\n472db13aaf docs: Address TODO\nb14f6ba62e Use NotificationFixture for legacy notifications too\nf1f599d098 Create a fixture around fake_notifier\n696fbab9e1 tests: Rename \u0027ImageBackendFixture\u0027 to \u0027LibvirtImageBackendFixture\u0027\ne2d25eb149 tests: Add os-brick fixture\n33be6d4b56 tests: Move libvirt-specific fixtures\nd0b8028b51 docs: Add releases to libvirt distro support matrix\n9d037f7d19 image_meta: Provide image_ref as the id when fetching from instance\nb28579a7b0 Describe the use of Review-Priority label\n0d84833e96 Use the new type HostDomainOpt.\n74724b4964 Change minversion of tox to 3.18.0\n781612b332 Reject open redirection in the console proxy\ne1cc0ed978 Add regression test for bug #1928063\n212f89a61e tests: Split external service fixtures out\nc269285568 tests: Move remaining non-libvirt fixtures\n2e53c82653 tests: Move ConfFixture\nd726ef51ec tests: Create \u0027nova.tests.fixtures\u0027 module\n150b918251 Replace getargspec with getfullargspec\n86af7feed0 Stop leaking ceph df cmd in RBD utils\n8b647f1b3f rbd: Get rbd_utils unit tests running again\n251e601a4d Fix typo in test_utils\n39a617752f Adapt to SQLAlchemy 1.4\n67f79fdb44 Add unit test for importing ed25519 ssh key\na62dd42c0d libvirt: Delegate OVS plug to os-vif\n0b05b838a4 Fix bond_mode enum 802.1ad -\u003e 802.3ad\ndab4ec1a53 zuul: Remove nova-dsvm-multinode-base\n91e53e4c2b zuul: Replace grenade and nova-grenade-multinode with grenade-multinode\n51d16adda6 mypy: Add type annotations to \u0027nova.pci\u0027\n00f1d4757e Update SRIOV port pci_slot when unshelving\n606d588e3e guestfs: With libguestfs \u003e\u003d v1.41.1 decode returned bytes to string\n7317cfbc67 Consolidate device detach error handling\n257b4d83d9 Move instance power state check to _detach_with_retry\nbfe50114e3 Drop support for SQLite \u003c 3.7\nbab3c8f3cb Remove references to \u0027sys.version_info\u0027\n3125d4c99c glance: Improve [glance]enable_rbd_download and associated options help text\nf8e68ddcd9 docs: Document location of nova.conf files\nb5edc294a1 docs: Add man pages for \u0027nova-policy\u0027\n0997ad15fc vmware: Use oslo.vmware\u0027s get_moref_value()\ndc814a6822 libvirt: Remove dead error handling code\na7f2b65d17 Follow up type hints for a634103\nc17f1e14cf Enable mypy on libvirt/guest.py\n52d6cd941c Move the guest.get_disk test to test_guest\ne56cc4f439 Replace blind retry with libvirt event waiting in detach\nbea06123db Test SRIOV port move operations with PCI conflicts\ncf4b152b6b vmware: Handle empty list attributes on vSphere objects\n07a9db5f74 Update min supported service version for Xena.\n6ce1272a62 vmware: OptionValue.value cannot be Boolean\n6bfa07b9d5 Explaining when a base image is considered unused to avoid confusion why some images aren\u0027t deleted.\n9e5feb051c docs: Populate \"See Also\" section with meaningful links\n99e5a1ac9a docs: Document options for remaining nova-* commands\na3c44c6243 [OVN] Explicitly set nova-next job to ML2/OVS\ne097c5a792 [OVN] Explicitly set grenade job to ML2/OVS\neaf6a828df Update contributor guide for Xena\n0a7d3794c6 libvirt: Ignore device already in the process of unplug errors\n9d285e0bb4 Switch to new rolevar for run-tempest role\n8e5515c1fa Trival change: spell error of Shelve\nd3d08922c7 Remove unused DeleteFromSelect db api method\nbf89a23d91 api: Reject requests to force up computes when `done` evacuation records exist\ndb333070b6 Add regression test for bug #1922053\n62868aaac7 Neutron fixture: don\u0027t clobber profile and vif_details if empty\n03ff08c7fe docs: Correct typos in \u0027nova-manage cells_v2\u0027 docs\n96bf8a0ab5 docs: Correct typos in \u0027nova-manage db\u0027 docs\nfe3a04be80 docs: Rewrite \u0027nova-manage libvirt\u0027 docs\n5f6aae3753 docs: Rewrite \u0027nova-manage placement\u0027 docs\n2f8ab439a7 docs: Rewrite \u0027nova-manage api_db\u0027, \u0027cell_v2\u0027 docs\ncefa8e0938 docs: Rewrite \u0027nova-manage db\u0027 docs to use proper directives\n59e2b38c80 docs: Use proper headers in nova-manage man page\n5c396c19cf docs: Follow up for SEV doc\ne5b264bf7b Bumping min os-brick ver to 4.3.1\ne19f6fda58 Follow up from bp/pci-socket-affinity series\nf35615e921 tests: Remove duplicate policy tests\n1748bd2509 tests: Remove useless mocks\n6dbc8e5416 tests: Merge flavor tests\n0ac74f4e00 Remove references to \u0027inst_type\u0027\n1bf45c4720 Remove (almost) all references to \u0027instance_type\u0027\n48429cc461 tests: Speed up \u0027servers\u0027 API tests\n1de6e960af Remove references to \u0027instance_type\u0027\n61256b4f78 Remove \u0027create_db_entry_for_new_instance\u0027\nb8ca4d3139 Add Python3 xena unit tests\n388498ac5f [neutron] Get only ID and name of the SGs from Neutron\n68ad2c67d8 Update master for stable/wallaby\n99409375a0 compute: Reject requests to commit intermediary snapshot of an inactive instance\ne89125a422 docs: Clarify value for \u0027hw:cpu_realtime\u0027 extra spec\n387a5753de doc: Use a non-numerical anchor when referencing microversions\n4908daed96 libvirt: Simplify device_path check in _detach_encryptor\n9dfac32959 docs: Add libvirt misc doc\n76549775fe docs: Change formatting of hypervisor config guides\n04b8693703 docs: Fold in MDS security flaw doc\nbe03ca7be7 docs: Clarify host-model, host-passthrough differences\nee1443bb3f docs: Add CPU models guide\n559404c23b trivial: fix word duplication in api ref\nefd2816619 zuul: Add nova-live-migration-ceph job\n4f1b879a68 docs: Change \u0027add_stylesheet\u0027 to \u0027add_css_file\u0027\n20a7c98eff [OVN] Adapt the live-migration job scripts to work with OVN\n9f260f1e6e Retry on vmware create_vm when it fails\n4dda84eb03 Placeholders for DB migration backports to Wallaby\nbe6927a28f console: Improve logging\ncf8ae56fab Enable mypy on scheduler/report.py\nde0e4e7e41 Turn on mypy for nova/network/neutron.py\n472825a839 vmware: Handle folder renames in rescue cmd\nbc0d1c8d63 nova-next: Drop NOVA_USE_SERVICE_TOKEN as it is now True by default\n2c19946d06 [ironic] correct capabilities for disk images supported\n5e2c31ab78 api: Add support for \u0027hostname\u0027 parameter\nb2781b98dd Functional tests removed direct post call\n4ccec62fd1 hardware: Add TODO to remove \u0027(un)pin_cpu_with_siblings\u0027\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I7aa3b297d766b6e7c05c1cfb4aad56b2fdaea42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c319d967f23df51e20b4c8fbe7b26cc8389352d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c319d967f23df51e20b4c8fbe7b26cc8389352d"}]},"branch":"refs/heads/master"},"19e579e42dcb71974a6566b335de81b5eb22df8f":{"kind":"REWORK","_number":2,"created":"2021-09-16 19:46:23.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/808706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/808706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/808706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/808706/2"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:30:41.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-16 19:46:15.000000000","tz":0},"subject":"Proposing Xena RC1 for nova","message":"Proposing Xena RC1 for nova\n\nThis proposed a RC1 release for nova to get any merged changes\nout prior to the RC1 deadline for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 16.\n\n$ git log --oneline --no-merges 23.0.0.0rc1..1c502ebaec\n73d78f01a0 Add the Xena prelude section\n70040856aa docs: Add nova-volume volume_attachment refresh admin workflow\na359753006 Fix nova-manage db version\n166b1730b8 Reproduce bug/1943436\n058d137ac8 Add more retries to TestMigrateFromDownHost tests\n22d4057a16 [doc] port-resource-request-groups not landed in Xena\nbbcd7d2fd0 fup: Print message logging uncaught nova-manage exceptions\na6ef502aa0 Support Cpu Compararion on Aarch64 Platform\nc590843974 Allow \u0027bochs\u0027 as a display device option\n8541f37cb5 Update compute rpc version alias for xena\nfc77ce191f Add --sleep option for archive_deleted_rows --until-complete\n625469d692 Avoid excessive sleep in vmware unit test\nb67b928c38 Parse alias from domain hostdev\n1b6a6e3916 Convert features not supported error to HTTPBadRequest\n37f80c13c6 [func test] move unshelve test to the proper place\n44309c419f Support interface attach / detach with new resource request format\n122a32ed82 workarounds: Remove rbd_volume_local_attach\n9bd62eae6e workarounds: Remove disable_native_luksv1\n9f8cc2f038 Add two new hacking rules\n9cb92e8832 [func test] refactor interface attach with qos\nfa35cfa998 tests: Address nits for configurable-instance-hostnames series\nc09d98dadb Add force kwarg to delete_allocation_for_instance\n6c5dd864c2 db: Handle parameters in DB strings\ncf07df57e7 Remove module level caching\n191bdf2069 Support move ops with extended resource request\n9be996c696 Expose the mdev class\nc3886c3ca7 Support boot with extended resource request\n2b847085c8 Provide and use other RCs for mdevs if needed\na543cb23aa Provide the mdev class for every PCI device\nfc1a99dc16 docs: admin/networking rename neutron_tunneled to neutron_tunnel\ncfa33d3b06 policy: Deprecate field from \u0027os-extended-server-attributes\u0027 policy\ne906a8c0ec nova-manage: Introduce volume show, refresh, get_connector commands\n191712594d Add some missing parameters in docs of os-cells\n38603a4d09 Reproduce bug 1941005\na057bbc406 fup: Fix os-volume_attachments api-ref parameters\n1d62485472 Fix documentation about cpu topologies\nac21c6674c api: Introduce microversion 2.89 adjusting os-volume_attachments\nb6ab1cb467 api: Log actual number of UUIDs queried\nda6af3fb22 compute: Query the service group API within check_instance_host\n5e8267f703 fup: Increase service_down_time beyond INITIAL_REPORTING_DELAY in test\n50a7a3050a conf: Remove deprecated aliases\n62e112461f tests: Merge \u0027test_utils\u0027, \u0027test_scheduler_utils\u0027\n6652999494 scheduler: Merge driver into manager\n289438b4c2 docs: Drop references to non-filter scheduler drivers\n930b7c9921 Avoid modifying the Mock class in test\n6fbd0b758d address open redirect with 3 forward slashes\n94f47471e0 Transfer RequestLevelParams from ports to scheduling\nd6cd4420bb Parse extended resource request from the port data\n9bae5f2ca5 Add missing [oslo_reports] options\nec6f8f63f9 [func test] ports with both bw and pps resources\nde9b136dc6 [func test] refactor asserts in qos tests\nc9b1dfe9b7 [func test] refactor assertPortMatchesAllocation\nb29ccf8aa5 Support same_subtree in allocation_canadidate query\nf6e8c512fb Bump min placement microversion to 1.36\n9de8bfb6e6 Add same_subtree field to RequestLevelParams\n8f5b7c67ca Reject server operations with extended resource req\ne357ad3c23 Reject server create with extended resource req\n017b0a3d23 Detect port-resource-request-groups neutron API extension\nf68d30bff3 [func test] neutron fixture for extended resource request\naa3d89b31f [func test] move port creation to the NeutronFixture\n713a11727d [func test] create pps resource on OVS agent RP\n1039aa6443 [func test] move port resource request tests\n2383cbb4a5 Fix inactive session error in compute node creation\na8492e8878 Prevent deletion of a compute node belonging to another host\nec48e1523d workarounds: Add libvirt_disable_apic\nb11e3f1d0d fup: Remove unused legacy block_device_info format\n78cace1ce4 fup: Move _migrate_stub into LibvirtMigrationMixin\n8a0ae569ef fixtures: Add legacy APIs to cinder fixture\ncb3e11952d tests: Sanity check some tests\ne3fa6859f9 api: Remove unnecessary \u0027base_\u0027 prefix from schemas\n0942163c82 api: Pass request to \u0027addresses\u0027 view builder\nb543f8226c api: Rename \u0027parameter_types.hostname\u0027 -\u003e \u0027fqdn\u0027\n3f548f3136 VmWare: Fix deprecation warning in unit tests\n1f2aa3e9d8 VMWare: Use get_hardware_devices throughout\n7fc6fe6fae Fix 1vcpu error with multiqueue and vif_type\u003dtap\nd31b26e09d VMWare: Use WithRetrieval to get all results\n39f0af5d18 libvirt: Abort live-migration job when monitoring fails\n20d8554f64 tests: Enable SADeprecationWarning warnings\neb728e877a db: Final cleanups\na7584ec1a5 docs: Add documentation on database migrations\n2226ec4e3b Accelerator smartnic SRIOV support\n2bb4527228 Invalidate provider tree when compute node disappears\n32676a9f45 Clear rebalanced compute nodes from resource tracker\n59d9871e8a Add functional regression test for bug 1853009\n31bedfc09c fup: Assert state of connection_info during LM rollback in func tests\nd4cc10d7bb fup: Make connection_info returned by CinderFixture unique per attachment\nce128798f5 fup: Mock all unix connectors in os-brick fixture\n79ec4e2938 fup: Move OSBrickFixture into base.ServersTestBase\n962eda94d5 compute: Ensure updates to bdms during pre_live_migration are saved\n268b716955 Add a regression test for bug 1939545\nd102b751b7 fake: Ensure need_legacy_block_device_info returns False\n1ba2c1c55d db: Enable auto-generation of migrations\n905c9723e9 db: Integrate alembic\n307adc5e08 db: Normalize migrations tests\nc3e113eb31 db: Trivial style changes\n38a022166f db: Add initial alembic migration for API DB\nca9ba88a7a db: Add initial alembic migration for main DB\n21c5ced0ba db: Avoid use of ALTER in initial migration\n43b253cd60 db: Post reshuffle cleanup\nbf8b5fc7d0 db: Move remaining \u0027nova.db.sqlalchemy\u0027 modules\naec2ca9468 db: Exclude the \u0027use_db_reconnect\u0027 option\n100b9dc62c db: Unify \u0027nova.db.api\u0027, \u0027nova.db.sqlalchemy.api\u0027\n84b6179076 db: Register database config options ourselves\n59c2262ca5 Block servers with vGPU and device profile in heal_allocations\nd4dbcd5fa0 func: Increase rpc_response_timeout in TestMultiCellMigrate tests\n4e18f07670 zuul: Mark live migration jobs as non-voting due to bug #1912310\n9cf07d5aaa Change the admin-visible logs for mdev support\nede7866de4 Add regression test for bug 1938326\nc3245098e3 smartnic support - functional tests\n1f53176d2f smartnic support - reject server move and suspend\ne19fa1a199 smartnic support - cleanup arqs\n120fe8af93 smartnic support - build instance with smartnic arqs\nb90c828d70 smartnic support - create arqs\n8b84135f2d smartnic support - new vnic type\n9c55d5d59d Smartnic support - cyborg drive\n2ffd973860 Add functional test for bug 1937375\n753a2ae5d4 fup: Move _wait_for_volume_attach into InstanceHelperMixin\n2209b0007f compute: Avoid duplicate BDMs during reserve_block_device_name\n49ba5a763f libvirt: Handle silent failures to extend volume within os-brick\nff4d0d002a Rename vgpu options to mdev\ndbf3e276ba zuul: Increase GLANCE_LIMIT_IMAGE_SIZE_TOTAL for nova-lvm\n39ef1ad774 Restore original time override in test_archive_task_logs\n214cabe684 Reduce mocking in test_reject_open_redirect for compat\na3d0aa4772 I2f9ad3df25306e070c8c3538bfed1212d6d8682f fup: add log\n48f35058fe tests: Remove \u0027stub_out_trusted_certs\u0027\nf916266d4f trivial: Cleanup a comment about a now removed libvirt version check\n880611df0b Allow deletion of compute service with no compute nodes\n32257a2a6d Reproducer unit test for bug 1860312\nd4f9b03c5c api: Improve extra spec validator help texts\n1d850a1593 use cinder v3\n809a52a16b Add tests for \u0027AggregateInstanceExtraSpecsFilter\u0027\nf0d17e774f Add tests for \u0027AggregateImagePropertiesIsolation\u0027\n89e321b7b5 tox: Encode specific Python versions\na65ea33824 Bump oslo.db to 10.0.0\n4ee2f667b5 tests: Validate AZ values\n52d9dd2c0b doc: allow_resize_to_same_host needs nova-api restarted\nd093849c3c Correct reno for archiving task_log database records\na22fe2da97 Bump os-resource-classes to 1.1.0\nad60f23be3 extend_volume of libvirt/volume/iscsi should not use device_path\n3bda965e15 tests: Remove unnecessary warnings filter\n0aea2ac903 tests: Silence SQLAlchemy warnings\n1d3373dcf0 Fix request path to query a resource provider by uuid\n982e2ee02d Use neutronclient\u0027s port binding APIs\n87dba379ac Make test_archive_task_logs deterministic\nbe9fe5ce61 Improve prep_resize reschedule unit test\nb84787d26a Add func test for nova-manage db archive_deleted_rows --before\nd88fe2895b docs: Replace \u0027nova boot\u0027 with \u0027openstack server create\u0027\n9cdecc81fb api: Handle invalid volume UUIDs during spawn\nfaa9de93d1 Remove the redundant code for attach interface\n8f21ee42bd api: Align availability zone info with forced host\n9799468d6f db: Drop support for experimental concurrency\n014480d52f db: Remove constant aliases from \u0027nova.db.api\u0027\n089ae234a7 db: Move \u0027sqlalchemy.types\u0027 up a directory\nbcf225daf4 db: Move main DB migrations\n4bbea58cbb db: Move db.sqalchemy.migration to db.migration\n43de2421b3 db: Use module-level imports for sqlalchemy (for real)\n97a566ad9c Add logs when cannot fit numa\ndfda0c0482 Fix oslo policy DeprecatedRule warnings\nf84d5917c6 [ironic] Minimize window for a resource provider to be lost\ne0534cc289 scheduler: Merge \u0027FilterScheduler\u0027 into base class\n6d745036bc trivial: Remove FakeScheduler (for realz)\nc4b4704783 scheduler: Remove \u0027hosts_up\u0027\nb20753b313 scheduler: \u0027USES_ALLOCATION_CANDIDATES\u0027 removal cleanup\n97d25371b3 scheduler: Remove \u0027USES_ALLOCATION_CANDIDATES\u0027\na569a51fed fix sr-iov support on Cavium ThunderX hosts.\n17cca270a6 Make explicit the network backend used in the CI jobs\nfa0fb2fe3d libvirt: Always delegate OVS plug to os-vif\nc31d3f96b1 Fix error \u0027404 Not Found\u0027\n99cf5292c7 objects: Fix VIFMigrateData.supports_os_vif_delegation setter\nb5b2e99824 zuul: Add nova-tox-functional-centos8-py36 job\n085de864db tests: Allow bindep and test-setup.sh to run on EL distros\n387823b36d Fix max cpu topologies with numa affinity\nfe25fa13e0 Test numa and vcpu topologies bug: #1910466\n4de5a92d4e db: Reintroduce validation of shadow table schema\nbc71331bca Add --task-log option to nova-manage db archive_deleted_rows\n98b01c9a59 Move \u0027check-cherry-picks\u0027 test to gate, n-v check\n070e83b749 db: Clean up migration code\nb5dd59f8f9 db: Synchronize function signatures\n1d60cd7e05 db: Copy docs from \u0027nova.db.*\u0027 to \u0027nova.db.sqlalchemy.*\u0027\n7ab2947720 db: Remove \u0027nova.db.base\u0027 module\n5819fe46a9 db: Fold in ForeignKey constraints\n66daa87339 db: Fold in indexes\n3b18b788f2 db: Use module-level imports for sqlalchemy\nd6cad899ca db: Remove unused DB methods\n194df1fa62 db: Remove \u0027nova.db.sqlalchemy.utils\u0027\n6df671fea7 db: Remove dead code\nd966546516 Add test coverage for API version headers in CORS\n6aa580c15f gate: Remove test_evacuate.sh\nb02a95a18b Allow X-OpenStack-Nova-API-Version header in CORS\ne59fc77c3d Retry lvm volume and volume group query\neb67ec24e3 Handle OPERATION_FAILED error during detach\n7c76821499 Add debug log for device detach libvirt error\n31491e5bc3 Fix typos in minimum version policy docs\n7c478ac099 zuul: Skip block migration with attached volumes tests due to bug #1931702\n47bedbb77a Revert \"Removing mypy to fix the nova CI\"\nd71dfa8b8d Move fake_notifier impl under NotificationFixture\n7085770b20 Test the NotificationFixture\n531fa4cd04 Removing mypy to fix the nova CI\n375a8452bb api: Log correct client IP if load balancer in use\n7a0b2d3482 docs: Explicitly state lack of support for node renaming\n5c02091647 requirements: Add types-paramiko\n4d8bf15fec libvirt: Set driver_iommu when attaching virtio devices to SEV instance\n26d65fc882 libvirt: Do not destroy volume secrets during _hard_reboot\nbfde686130 Remove unnecessary mocks from unit test\n7b11409463 Make test_refresh_associations_* deterministic\n0b63f9effe Improve policy doc for supported scope info\n2c761066e0 Remove PROJECT_ADMIN limitation from zero-disk and external-network policy\n1cc52fd2f2 virt: Add destroy_secrets kwarg to destroy and cleanup\n33c8af1f8c Error anti-affinity violation on migrations\n7c7a2a142d Deprecate filters that have been replaced by placement filters\n5cf06bf33d docs: Remove duplicate ToC entry\nfc3838a9b3 Change nova doc after Freenode -\u003e OFTC move\n5c50a45a1e Fix RequestLevelParams persistence handling in RequestSpec\n17ac1a40cb Add test coverage for security group checking\n56eb253e9f Honor [neutron]http_retries in the manual client\ne98994027f hardware: Use image_meta.id within get_mem_encryption_constraint\nd5ed968826 zuul: Skip swap_volume tests as part of nova-next\n740e6f09bf libvirt: Enable \u0027vmcoreinfo\u0027 feature by default\n472db13aaf docs: Address TODO\nb14f6ba62e Use NotificationFixture for legacy notifications too\nf1f599d098 Create a fixture around fake_notifier\n696fbab9e1 tests: Rename \u0027ImageBackendFixture\u0027 to \u0027LibvirtImageBackendFixture\u0027\ne2d25eb149 tests: Add os-brick fixture\n33be6d4b56 tests: Move libvirt-specific fixtures\nd0b8028b51 docs: Add releases to libvirt distro support matrix\n9d037f7d19 image_meta: Provide image_ref as the id when fetching from instance\nb28579a7b0 Describe the use of Review-Priority label\n0d84833e96 Use the new type HostDomainOpt.\n74724b4964 Change minversion of tox to 3.18.0\n781612b332 Reject open redirection in the console proxy\ne1cc0ed978 Add regression test for bug #1928063\n212f89a61e tests: Split external service fixtures out\nc269285568 tests: Move remaining non-libvirt fixtures\n2e53c82653 tests: Move ConfFixture\nd726ef51ec tests: Create \u0027nova.tests.fixtures\u0027 module\n150b918251 Replace getargspec with getfullargspec\n86af7feed0 Stop leaking ceph df cmd in RBD utils\n8b647f1b3f rbd: Get rbd_utils unit tests running again\n251e601a4d Fix typo in test_utils\n39a617752f Adapt to SQLAlchemy 1.4\n67f79fdb44 Add unit test for importing ed25519 ssh key\na62dd42c0d libvirt: Delegate OVS plug to os-vif\n0b05b838a4 Fix bond_mode enum 802.1ad -\u003e 802.3ad\ndab4ec1a53 zuul: Remove nova-dsvm-multinode-base\n91e53e4c2b zuul: Replace grenade and nova-grenade-multinode with grenade-multinode\n51d16adda6 mypy: Add type annotations to \u0027nova.pci\u0027\n00f1d4757e Update SRIOV port pci_slot when unshelving\n606d588e3e guestfs: With libguestfs \u003e\u003d v1.41.1 decode returned bytes to string\n7317cfbc67 Consolidate device detach error handling\n257b4d83d9 Move instance power state check to _detach_with_retry\nbfe50114e3 Drop support for SQLite \u003c 3.7\nbab3c8f3cb Remove references to \u0027sys.version_info\u0027\n3125d4c99c glance: Improve [glance]enable_rbd_download and associated options help text\nf8e68ddcd9 docs: Document location of nova.conf files\nb5edc294a1 docs: Add man pages for \u0027nova-policy\u0027\n0997ad15fc vmware: Use oslo.vmware\u0027s get_moref_value()\ndc814a6822 libvirt: Remove dead error handling code\na7f2b65d17 Follow up type hints for a634103\nc17f1e14cf Enable mypy on libvirt/guest.py\n52d6cd941c Move the guest.get_disk test to test_guest\ne56cc4f439 Replace blind retry with libvirt event waiting in detach\nbea06123db Test SRIOV port move operations with PCI conflicts\ncf4b152b6b vmware: Handle empty list attributes on vSphere objects\n07a9db5f74 Update min supported service version for Xena.\n6ce1272a62 vmware: OptionValue.value cannot be Boolean\n6bfa07b9d5 Explaining when a base image is considered unused to avoid confusion why some images aren\u0027t deleted.\n9e5feb051c docs: Populate \"See Also\" section with meaningful links\n99e5a1ac9a docs: Document options for remaining nova-* commands\na3c44c6243 [OVN] Explicitly set nova-next job to ML2/OVS\ne097c5a792 [OVN] Explicitly set grenade job to ML2/OVS\neaf6a828df Update contributor guide for Xena\n0a7d3794c6 libvirt: Ignore device already in the process of unplug errors\n9d285e0bb4 Switch to new rolevar for run-tempest role\n8e5515c1fa Trival change: spell error of Shelve\nd3d08922c7 Remove unused DeleteFromSelect db api method\nbf89a23d91 api: Reject requests to force up computes when `done` evacuation records exist\ndb333070b6 Add regression test for bug #1922053\n62868aaac7 Neutron fixture: don\u0027t clobber profile and vif_details if empty\n03ff08c7fe docs: Correct typos in \u0027nova-manage cells_v2\u0027 docs\n96bf8a0ab5 docs: Correct typos in \u0027nova-manage db\u0027 docs\nfe3a04be80 docs: Rewrite \u0027nova-manage libvirt\u0027 docs\n5f6aae3753 docs: Rewrite \u0027nova-manage placement\u0027 docs\n2f8ab439a7 docs: Rewrite \u0027nova-manage api_db\u0027, \u0027cell_v2\u0027 docs\ncefa8e0938 docs: Rewrite \u0027nova-manage db\u0027 docs to use proper directives\n59e2b38c80 docs: Use proper headers in nova-manage man page\n5c396c19cf docs: Follow up for SEV doc\ne5b264bf7b Bumping min os-brick ver to 4.3.1\ne19f6fda58 Follow up from bp/pci-socket-affinity series\nf35615e921 tests: Remove duplicate policy tests\n1748bd2509 tests: Remove useless mocks\n6dbc8e5416 tests: Merge flavor tests\n0ac74f4e00 Remove references to \u0027inst_type\u0027\n1bf45c4720 Remove (almost) all references to \u0027instance_type\u0027\n48429cc461 tests: Speed up \u0027servers\u0027 API tests\n1de6e960af Remove references to \u0027instance_type\u0027\n61256b4f78 Remove \u0027create_db_entry_for_new_instance\u0027\nb8ca4d3139 Add Python3 xena unit tests\n388498ac5f [neutron] Get only ID and name of the SGs from Neutron\n68ad2c67d8 Update master for stable/wallaby\n99409375a0 compute: Reject requests to commit intermediary snapshot of an inactive instance\ne89125a422 docs: Clarify value for \u0027hw:cpu_realtime\u0027 extra spec\n387a5753de doc: Use a non-numerical anchor when referencing microversions\n4908daed96 libvirt: Simplify device_path check in _detach_encryptor\n9dfac32959 docs: Add libvirt misc doc\n76549775fe docs: Change formatting of hypervisor config guides\n04b8693703 docs: Fold in MDS security flaw doc\nbe03ca7be7 docs: Clarify host-model, host-passthrough differences\nee1443bb3f docs: Add CPU models guide\n559404c23b trivial: fix word duplication in api ref\nefd2816619 zuul: Add nova-live-migration-ceph job\n4f1b879a68 docs: Change \u0027add_stylesheet\u0027 to \u0027add_css_file\u0027\n20a7c98eff [OVN] Adapt the live-migration job scripts to work with OVN\n9f260f1e6e Retry on vmware create_vm when it fails\n4dda84eb03 Placeholders for DB migration backports to Wallaby\nbe6927a28f console: Improve logging\ncf8ae56fab Enable mypy on scheduler/report.py\nde0e4e7e41 Turn on mypy for nova/network/neutron.py\n472825a839 vmware: Handle folder renames in rescue cmd\nbc0d1c8d63 nova-next: Drop NOVA_USE_SERVICE_TOKEN as it is now True by default\n2c19946d06 [ironic] correct capabilities for disk images supported\n5e2c31ab78 api: Add support for \u0027hostname\u0027 parameter\nb2781b98dd Functional tests removed direct post call\ndb524ef74d Address release note nits for cpu-resources series\n4ccec62fd1 hardware: Add TODO to remove \u0027(un)pin_cpu_with_siblings\u0027\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I7aa3b297d766b6e7c05c1cfb4aad56b2fdaea42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19e579e42dcb71974a6566b335de81b5eb22df8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19e579e42dcb71974a6566b335de81b5eb22df8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
