)]}'
{"id":"openstack%2Freleases~808710","triplet_id":"openstack%2Freleases~master~I957e0e1c426b6f6b6a98f5cc8f1f050531177a37","project":"openstack/releases","branch":"master","topic":"xena-rc1-deadline","hashtags":[],"change_id":"I957e0e1c426b6f6b6a98f5cc8f1f050531177a37","subject":"Proposing Xena RC1 for octavia","status":"MERGED","created":"2021-09-13 09:53:04.000000000","updated":"2021-09-17 16:02:18.000000000","submitted":"2021-09-17 16:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808710-xena-rc1-deadline","meta_rev_id":"8314ec1328859bdaf21dfeba565276c5ff8e7657","_number":808710,"virtual_id_number":808710,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 16:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-15 17:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-17 15:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-16 14:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2021-09-15 17:09:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-17 15:40:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-15 14:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 09:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-13 10:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 11:21:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-13 11:21:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-13 11:21:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-15 17:08:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"df2cbced651dc507fe63472760206fa0e96b9e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 09:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28c080e33a8ce2454810436a23cd5d060bfeaa28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 10:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a136434daf4d34bd8b461bdb123791 : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac55345fd8c94c5aac04f5ce828d1a27 : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/694e9d7bbe0a4401b9df1d3ca9d582c1 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a96c3f225c760d6870ee6edc7807b81ffa699bed","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-14 15:03:32.000000000","message":"Patch Set 1: Code-Review-1\n\nWaiting for 1 or 2 patches that can still make it for RC1","accounts_in_message":[],"_revision_number":1},{"id":"95bc6c6ae0cefa46bdee82be77ae642707b10284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 14:26:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee7a4e1cb787ff16ff8d22f1040e65474eb34ebe","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:28:20.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3b275caf147cd664a185985fd947d9b9d5cc35f5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 14:30:20.000000000","message":"Patch Set 2:\n\nupdated to current master (3 patches added)","accounts_in_message":[],"_revision_number":2},{"id":"81503c1cc314950922b93fd0d8387497a68fcfce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:35:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd51feb83c06401cb37a7b45958de568 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8c8950adc3a413abfec3000d1f8c4bc : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f4289df68804660a1223502a81ce45b : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c9888bed14d39b0fe17d303e1c783afd56605344","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-15 17:08:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* MAJOR version bump is required\n* updated by the PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":2},{"id":"5385ac5a788f6a676387334bca0072112b1c325d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-15 17:09:05.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"31d0c37ecfac1d34805ee239f8e3f2dbc2e36f5c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-16 14:55:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0eb211d597637a28569601a92d79e0fd0aa3f46a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 15:40:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dbdbcf97675b40b0845389d04a0babfb3ee6bc6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 15:41:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b95bd790565d8ffdb2361226db0db106ecaab6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:02:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de8dd69864f4d6cab59b2c7bc80d5f1 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f8753d353e4a509ee25193f7f4dfb5 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/793c3ec9eb0e4fee9209b5c084d6b7e5 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8314ec1328859bdaf21dfeba565276c5ff8e7657","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:02:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a284b9771c58a94ee2094a9ad18c226f807def8","revisions":{"43217d7e9e9b5f3b168186bd606e904b0b5357e5":{"kind":"REWORK","_number":1,"created":"2021-09-13 09:53:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/808710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/808710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/808710/1"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:52:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:52:56.000000000","tz":120},"subject":"Proposing Xena RC1 for octavia","message":"Proposing Xena RC1 for octavia\n\nThis proposed a RC1 release for octavia to get any merged changes\nout prior to the RC1 deadline for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 16.\n\n$ git log --oneline --no-merges 8.0.0.0rc1..69d975d6\nb88154f0 Fix PlugVIPAmphora revert function in amphorav2\n5dd7ad9a Add generic network interface management in the amphora\n39735ebf Set active-standby job as non-voting\n20da8e1a Fix pylint checks\n1aef468e Alias change amphorav2 -\u003e amphora\n4174f4a5 Update nr_open limit value in the amphora\nb89c929c Add caps for pip-extra-reqs/pip-missing-reqs\n7e264e1f Fix oslo policy DeprecatedRule warnings\ndb7a633a Replace md5 for fips\n4a9de5b2 Fix new pylint issues\n1b365ba4 Changed minversion in tox to 3.18.0\n76606b4c Fix using host_routes in VIP subnet with amphorav2\n4443596e Enable lo interface in the amphora-haproxy netns\nc58a2c7c Fix new errors with SQLAlchemy 1.4\n2bcb344f Update contributor documentation\ne1adb335 Imported Translations from Zanata\n0a5e7ef9 Drop Bionic support\na26494bf Edits for \"Basic Load Balancing Cookbook\"\ncd272ae5 Fix race conditions between API and worker DB calls\n7075d22c Optimize CountPoolChildrenForQuota task in amphorav2\n815a2838 Spare pool removal\nb2f49250 setup.cfg: Replace dashes with underscores\nc5287e1b Fix task_flow.max_workers with persistence in amphorav2\n4e53676b Fix pylint errors\n56f59878 Use TOX_CONSTRAINTS_FILE\nc1910f37 Add nightly test image build job for focal\n190d6a33 Fix jobboard_enabled setting in devstack\n042a6ea9 Disable spare pool jobs\n7e2dab64 Fix amphora failover flow in amphorav2 driver\nbeccbb17 Gate-Fix: Replace Octavia hacking O349 with H216\nde352cac Edits for \"Operator Maintenance Guide\"\nee0da827 Fix comment for the ca_certificates_file opt\n5d470357 Fix devstack cleanup when using amphorav2\na8e3f402 Fix pool ALPN compatibility with older amphora\ne26664de Fix rsyslog configuration when disabling logs\n380f5c4c Update master for stable/wallaby\n8d86187c Validate user access to vip_subnet_id when creating a LB\n40d011ab Make focal the Ubuntu default for diskimage-create\n5352a10f Fix weighted round-robin about UDP and SCTP listener\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I957e0e1c426b6f6b6a98f5cc8f1f050531177a37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/43217d7e9e9b5f3b168186bd606e904b0b5357e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/43217d7e9e9b5f3b168186bd606e904b0b5357e5"}]},"branch":"refs/heads/master"},"7a284b9771c58a94ee2094a9ad18c226f807def8":{"kind":"REWORK","_number":2,"created":"2021-09-15 14:26:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/808710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/808710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/808710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/808710/2"}}},"commit":{"parents":[{"commit":"25b9dad49e73bda877ca0579ea1da00f0f764915","subject":"Merge \"[blazar] Create for all client and non-client lib xena branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25b9dad49e73bda877ca0579ea1da00f0f764915"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-13 09:52:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-15 14:22:54.000000000","tz":120},"subject":"Proposing Xena RC1 for octavia","message":"Proposing Xena RC1 for octavia\n\nThis proposed a RC1 release for octavia to get any merged changes\nout prior to the RC1 deadline for Xena going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, September 16.\n\n$ git log --oneline --no-merges 8.0.0.0rc1..b8ddbc94\n3a8f0563 Fix MAX_TIMEOUT value for listener\n151a9432 Fix duplicate SG creation for listener peer port\nb88154f0 Fix PlugVIPAmphora revert function in amphorav2\n5dd7ad9a Add generic network interface management in the amphora\n39735ebf Set active-standby job as non-voting\n20da8e1a Fix pylint checks\n1aef468e Alias change amphorav2 -\u003e amphora\n4174f4a5 Update nr_open limit value in the amphora\nb89c929c Add caps for pip-extra-reqs/pip-missing-reqs\n7e264e1f Fix oslo policy DeprecatedRule warnings\ndb7a633a Replace md5 for fips\n4a9de5b2 Fix new pylint issues\n1b365ba4 Changed minversion in tox to 3.18.0\n76606b4c Fix using host_routes in VIP subnet with amphorav2\n4443596e Enable lo interface in the amphora-haproxy netns\nc58a2c7c Fix new errors with SQLAlchemy 1.4\n2bcb344f Update contributor documentation\nb3f59eeb Disable conntrack for TCP flows in the amphora\ne1adb335 Imported Translations from Zanata\n0a5e7ef9 Drop Bionic support\na26494bf Edits for \"Basic Load Balancing Cookbook\"\ncd272ae5 Fix race conditions between API and worker DB calls\n7075d22c Optimize CountPoolChildrenForQuota task in amphorav2\n815a2838 Spare pool removal\nb2f49250 setup.cfg: Replace dashes with underscores\nc5287e1b Fix task_flow.max_workers with persistence in amphorav2\n4e53676b Fix pylint errors\n56f59878 Use TOX_CONSTRAINTS_FILE\nc1910f37 Add nightly test image build job for focal\n190d6a33 Fix jobboard_enabled setting in devstack\n042a6ea9 Disable spare pool jobs\n7e2dab64 Fix amphora failover flow in amphorav2 driver\nbeccbb17 Gate-Fix: Replace Octavia hacking O349 with H216\nde352cac Edits for \"Operator Maintenance Guide\"\nee0da827 Fix comment for the ca_certificates_file opt\n5d470357 Fix devstack cleanup when using amphorav2\na8e3f402 Fix pool ALPN compatibility with older amphora\ne26664de Fix rsyslog configuration when disabling logs\n380f5c4c Update master for stable/wallaby\n8d86187c Validate user access to vip_subnet_id when creating a LB\n40d011ab Make focal the Ubuntu default for diskimage-create\n5352a10f Fix weighted round-robin about UDP and SCTP listener\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I957e0e1c426b6f6b6a98f5cc8f1f050531177a37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a284b9771c58a94ee2094a9ad18c226f807def8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a284b9771c58a94ee2094a9ad18c226f807def8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
