)]}'
{"id":"openstack%2Freleases~809180","triplet_id":"openstack%2Freleases~master~Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65","project":"openstack/releases","branch":"master","topic":"stein-eol","hashtags":[],"change_id":"Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65","subject":"[vpnaas] Transition stein, rocky, queens, pike to EOL","status":"MERGED","created":"2021-09-15 12:43:02.000000000","updated":"2021-10-04 09:59:37.000000000","submitted":"2021-10-04 09:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809180-stein-eol","meta_rev_id":"ce87514c08b78f96ce9c0cafd4b30dabca7700bb","_number":809180,"virtual_id_number":809180,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 09:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-10-04 09:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-27 11:09:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-10-04 09:05:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-10-04 09:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-10-04 09:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-10-04 09:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 12:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-15 12:47:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-15 12:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 13:20:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-09-15 14:45:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-16 11:42:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"state":"REVIEWER"},{"updated":"2021-09-16 11:42:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"state":"REVIEWER"},{"updated":"2021-09-20 07:15:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-27 08:20:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-10-04 09:48:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"02d4fd253e5c557413f0c81bdf5476bd399386ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-15 12:43:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9b82a3902cc81e17c1ac4b0ffe3a2e4b7d4970d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-15 12:47:50.000000000","message":"Patch Set 1: Code-Review+1\n\nIn favour of this, per last backport dates in [2] and current CI status in these older branches - all red, with quite some work required to get them back in order, see https://review.opendev.org/c/openstack/neutron-vpnaas/+/805969 for train fixes needed","accounts_in_message":[],"_revision_number":1},{"id":"d2b79ad09bfa98edaee75d5587890c1c3d5f792b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 12:51:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7473285dfb49437a96279bf3b9899c5a : FAILURE in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc1edf380b604105849bb9ec36d04efd : FAILURE in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5645fd3759b04baa82d2da738a559230 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a4bff8e88eed470b8b128d782af081295a8dcf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-15 13:50:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"36b3702868035ad6ef6290ccd884092d209a0c5d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 13:52:30.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"08df1a36053f184b54568f7137b25add5296a910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:02:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41223abe53e4c989d152e6f148b7698 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc2b1c90cba1490db3dd98e0ef1819bd : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/467ce6ad1dc54d50a93d6746700b2736 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"417016b25bf6b2a96b384d1439bd007a1b13f6de","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-15 14:36:56.000000000","message":"Patch Set 2: Code-Review+1\n\nSneaky whitespace!","accounts_in_message":[],"_revision_number":2},{"id":"318eca5831a9d61aeaa4bc8ed4de31e79197458d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-15 14:45:53.000000000","message":"Patch Set 2: Code-Review-1\n\nIt seems that pike is also open, could you please add that to this patch as well? Thanks!\n\n(Similarly like you did with ocata: https://review.opendev.org/c/openstack/releases/+/799473/3/deliverables/ocata/neutron-vpnaas.yaml )","accounts_in_message":[],"_revision_number":2},{"id":"1ef6ac5492e1b3910113315a1cec9a2cf4c91a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-16 11:01:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7a55e5175c5df8bfa158d6845fcfd7f82f9735b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-16 11:03:05.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e It seems that pike is also open, could you please add that to this patch as well? Thanks!\n\u003e \n\u003e (Similarly like you did with ocata: https://review.opendev.org/c/openstack/releases/+/799473/3/deliverables/ocata/neutron-vpnaas.yaml )\n\nThanks, I added pike, but only for neutron-vpnaas not for dashboard","accounts_in_message":[],"_revision_number":3},{"id":"25715cb24c3b07926ada61403fd44e33c35f257f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 11:12:54.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"093377f86771a0b94256aec6c8d9dc2c82c30abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 11:22:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb326cd818c497ba02f8405794fc778 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a626738b90154d4c9e1300d994eda78e : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/088343c1083e45a2975009e4efd3d11a : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1cd18390ee33c3bd1cdb7904abaf44e04bec7f0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-16 12:20:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"40518498b0f4a1269c5e3c9529733bc0c8d1db76","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:40:44.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"e0168abe4624a23d5703a7005d14cb52f8f8d217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:21:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15ee940768841218602ba8b759bb096 : TIMED_OUT in 31m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16f4a00eba2d450fa0c62f5e24d775aa : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3a0d86c6cdfc46d690e2bbfa601758a6 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":4},{"id":"8e59a20728ed2b3a7ba37cd26ef96acf9d8d5a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-16 13:45:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"305a921236b50400df35ceb2f7022f015935f836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:49:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee8f3ca09aa4f94ba57948a1fdd42b9 : TIMED_OUT in 31m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e84660ca2a94693bd7303b5073d2d4e : TIMED_OUT in 1h 01m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/acc099405b3542ee95a76be2a62825e7 : TIMED_OUT in 1h 01m 10s","accounts_in_message":[],"_revision_number":5},{"id":"fcd98c644f23acf829c9ee3cc5506276feee5a0d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-17 09:55:48.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"397a4b315e1e2b431cf7cd0ddcb6dff852a143a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 11:15:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a76ab1ee22b4045ba0b22495fa89446 : TIMED_OUT in 31m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/852891264b28460c8ef1b50e76ca2ae2 : TIMED_OUT in 1h 01m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/620355f8fcc04dda99df975dd5ff78c0 : TIMED_OUT in 1h 01m 17s","accounts_in_message":[],"_revision_number":5},{"id":"6931a712df5c9dbea4d73d5a17531b2751522219","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-17 12:53:01.000000000","message":"Patch Set 5:\n\nrecheck\nBased on release channel\u0027s discussion the gate is free","accounts_in_message":[],"_revision_number":5},{"id":"2d061629f207684869f4d0c611b75abd28ef8b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:20:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7f8f1aee2e4b0981089998a73a7a84 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/863342b59e3c451cb987a13cbf8e3853 : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ea7dd2fa321e44d6a03a78daa7f90010 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":5},{"id":"70596237f3269377f929160363e871e34ab40dd4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-17 16:25:08.000000000","message":"Patch Set 5:\n\nShould we get https://review.opendev.org/c/openstack/neutron-vpnaas/+/809185 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809186 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809183 in and use the new hashes? to drop periodic jobs. Or will this be done with EOL process?","accounts_in_message":[],"_revision_number":5},{"id":"ee2b074e67922f10af7e14ca149107e24325159c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-27 08:20:51.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e Should we get https://review.opendev.org/c/openstack/neutron-vpnaas/+/809185 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809186 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809183 in and use the new hashes? to drop periodic jobs. Or will this be done with EOL process?\n\nIndeed these patches could be included before moving to EOL. The branches will be deleted after eoling the deliverables so if these patches are not merged before then they will be lost. Merging them first will backup them in the eol tag.","accounts_in_message":[],"_revision_number":5},{"id":"4846b639f41b54a901e17e434652812d2dda75bf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-27 08:45:28.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e Should we get https://review.opendev.org/c/openstack/neutron-vpnaas/+/809185 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809186 https://review.opendev.org/c/openstack/neutron-vpnaas/+/809183 in and use the new hashes? to drop periodic jobs. Or will this be done with EOL process?\n\u003e \n\u003e Indeed these patches could be included before moving to EOL. The branches will be deleted after eoling the deliverables so if these patches are not merged before then they will be lost. Merging them first will backup them in the eol tag.\n\nSorry for the late answer and thanks Hervé for the answer. To add some more detail for the answer:\n\n* it is not necessary to delete zuul jobs from .zuul.yaml on the branch that will be EOL\u0027d, because as soon as the branch gets deleted zuul won\u0027t pick up any jobs from there\n* it is NECESSARY to delete zuul job definitions that are defined in OTHER repository than neutron-vpnaas AND won\u0027t be used anymore (i.e. the job is only used in the to-be-deleted branches of neutron-vpnaas), so that we won\u0027t have \u0027dead code\u0027 among zuul job definitions\n* to be able to delete a branch after it is tagged as EOL all open patches should be abandoned\n\nTL;DR: we don\u0027t need those job dropping patches to be merged","accounts_in_message":[],"_revision_number":5},{"id":"da6739fcf31afe727219ad91e4714dcdac7d9cb1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-27 11:09:56.000000000","message":"Patch Set 5: Code-Review+1\n\nThanks Elod and Hervé, so this looks good to go in then! (and then clean the mentioned patches to actaully delete the branch)","accounts_in_message":[],"_revision_number":5},{"id":"51c88abd26ff77f9872ffcd4f0803fade597a331","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-04 09:05:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"61afe4054192e343fce05154ec8002fea5141b01","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:06:57.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"025ec7ba97ddcf09c54b4687b49ff7347f8dabb0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-04 09:15:04.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM:\n* mail was sent to ML almost a month ago\n* hashes look correct (all from head of the given stable/* branch)\n* approved by PTL","accounts_in_message":[],"_revision_number":5},{"id":"14fb2e5276b2a23409c3aca3bd0b277ff2831b57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-04 09:48:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"42d7191288f1c6628417364a6593c5b84a98bf9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:50:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2eca30fe5b1ee92a03ede5a538d58d43bcf66836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:59:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa1b76acf8d4d2ca745c62a68d741e6 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6875280f4ab1455985f03757290cb466 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bda1dc7e68f14aeca98fa3030b156198 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5},{"id":"ce87514c08b78f96ce9c0cafd4b30dabca7700bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7a7b4a9ecdce2033ba13053695d69410d1be301e","revisions":{"84cfb541943e1d948a5c8fdfc2b0db690fdb0b23":{"kind":"REWORK","_number":1,"created":"2021-09-15 12:43:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/809180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/809180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/809180/1"}}},"commit":{"parents":[{"commit":"f182bef4ad74f208f6f42063deb28f6a51745ad9","subject":"Merge \"Release Glance stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f182bef4ad74f208f6f42063deb28f6a51745ad9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"subject":"[vpnaas] Transition stein, rocky, queens to EOL","message":"[vpnaas] Transition stein, rocky, queens to EOL\n\nThis patch transitions stable/stein, stable/rocky and stable/queens\nbranches to End of Life. This was discussed on Neutron CI meeting\n(see [1]) and a mail was sent out to the community (see [2]).\nThese branches cannot be used anymore and will be deleted if this patch\nmerges.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-09-07.log.html#t2021-09-07T15:18:43\n[2]: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024779.html\n\nChange-Id: Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84cfb541943e1d948a5c8fdfc2b0db690fdb0b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84cfb541943e1d948a5c8fdfc2b0db690fdb0b23"}]},"branch":"refs/heads/master"},"da398b995ccd66e85f06dca1ac8c6f0fd411c18f":{"kind":"REWORK","_number":2,"created":"2021-09-15 13:50:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/809180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/809180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/809180/2"}}},"commit":{"parents":[{"commit":"f182bef4ad74f208f6f42063deb28f6a51745ad9","subject":"Merge \"Release Glance stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f182bef4ad74f208f6f42063deb28f6a51745ad9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-09-15 13:50:54.000000000","tz":0},"subject":"[vpnaas] Transition stein, rocky, queens to EOL","message":"[vpnaas] Transition stein, rocky, queens to EOL\n\nThis patch transitions stable/stein, stable/rocky and stable/queens\nbranches to End of Life. This was discussed on Neutron CI meeting\n(see [1]) and a mail was sent out to the community (see [2]).\nThese branches cannot be used anymore and will be deleted if this patch\nmerges.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-09-07.log.html#t2021-09-07T15:18:43\n[2]: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024779.html\n\nChange-Id: Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da398b995ccd66e85f06dca1ac8c6f0fd411c18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da398b995ccd66e85f06dca1ac8c6f0fd411c18f"}]},"branch":"refs/heads/master"},"a06bf86b3dacf38ebfe5d07b4d4ab6279ebc5a8e":{"kind":"REWORK","_number":3,"created":"2021-09-16 11:01:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/809180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/809180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/809180/3"}}},"commit":{"parents":[{"commit":"f182bef4ad74f208f6f42063deb28f6a51745ad9","subject":"Merge \"Release Glance stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f182bef4ad74f208f6f42063deb28f6a51745ad9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-16 11:01:03.000000000","tz":120},"subject":"[vpnaas] Transition stein, rocky, queens, pike to EOL","message":"[vpnaas] Transition stein, rocky, queens, pike to EOL\n\nThis patch transitions stable/stein, stable/rocky, stable/queens\nand stable/pike branches to End of Life. This was discussed on Neutron\nCI meeting (see [1]) and a mail was sent out to the community (see [2]).\nThese branches cannot be used anymore and will be deleted if this patch\nmerges.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-09-07.log.html#t2021-09-07T15:18:43\n[2]: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024779.html\n\nChange-Id: Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a06bf86b3dacf38ebfe5d07b4d4ab6279ebc5a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a06bf86b3dacf38ebfe5d07b4d4ab6279ebc5a8e"}]},"branch":"refs/heads/master"},"db38cc6ba0f8be24e9dbe235a9d558295572fedc":{"kind":"REWORK","_number":4,"created":"2021-09-16 12:20:50.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/809180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/809180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/809180/4"}}},"commit":{"parents":[{"commit":"f182bef4ad74f208f6f42063deb28f6a51745ad9","subject":"Merge \"Release Glance stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f182bef4ad74f208f6f42063deb28f6a51745ad9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-16 12:20:29.000000000","tz":120},"subject":"[vpnaas] Transition stein, rocky, queens, pike to EOL","message":"[vpnaas] Transition stein, rocky, queens, pike to EOL\n\nThis patch transitions stable/stein, stable/rocky, stable/queens\nand stable/pike branches to End of Life. This was discussed on Neutron\nCI meeting (see [1]) and a mail was sent out to the community (see [2]).\nThese branches cannot be used anymore and will be deleted if this patch\nmerges.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-09-07.log.html#t2021-09-07T15:18:43\n[2]: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024779.html\n\nChange-Id: Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db38cc6ba0f8be24e9dbe235a9d558295572fedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db38cc6ba0f8be24e9dbe235a9d558295572fedc"}]},"branch":"refs/heads/master"},"7a7b4a9ecdce2033ba13053695d69410d1be301e":{"kind":"REWORK","_number":5,"created":"2021-09-16 13:45:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/809180/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/809180/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/809180/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/809180/5"}}},"commit":{"parents":[{"commit":"f182bef4ad74f208f6f42063deb28f6a51745ad9","subject":"Merge \"Release Glance stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f182bef4ad74f208f6f42063deb28f6a51745ad9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-15 12:32:34.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-09-16 13:45:11.000000000","tz":120},"subject":"[vpnaas] Transition stein, rocky, queens, pike to EOL","message":"[vpnaas] Transition stein, rocky, queens, pike to EOL\n\nThis patch transitions stable/stein, stable/rocky, stable/queens\nand stable/pike branches to End of Life. This was discussed on Neutron\nCI meeting (see [1]) and a mail was sent out to the community (see [2]).\nThese branches cannot be used anymore and will be deleted if this patch\nmerges.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-09-07.log.html#t2021-09-07T15:18:43\n[2]: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/024779.html\n\nChange-Id: Ic7ed1133f81d1637429aa7a69eaa7e799ed5bd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a7b4a9ecdce2033ba13053695d69410d1be301e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a7b4a9ecdce2033ba13053695d69410d1be301e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
