)]}'
{"id":"openstack%2Freleases~809419","triplet_id":"openstack%2Freleases~master~I56755457df76321a503ae5cc86690215ce4168a1","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I56755457df76321a503ae5cc86690215ce4168a1","subject":"Release openstackdocstheme 2.3.1","status":"MERGED","created":"2021-09-16 13:59:48.000000000","updated":"2021-09-20 15:07:22.000000000","submitted":"2021-09-20 15:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809419","meta_rev_id":"6dbec4ace7b992853b3294d726042093603a13a3","_number":809419,"virtual_id_number":809419,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 15:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-09-20 09:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-09-20 14:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-16 15:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-09-20 14:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-16 14:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 14:06:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-16 15:54:45.000000000","updated_by":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"reviewer":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"state":"REVIEWER"},{"updated":"2021-09-20 09:08:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-09-20 14:53:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"f49271fdc0f813799df60c9b15306e1294e49a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-16 13:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d87ce315f4c5dd0212bb5970d7b6fbc84f5807b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-16 14:02:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a18d672f1721b1e8e531637b4356a62ff0a34b28","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-16 14:03:17.000000000","message":"Topic openstackdocstheme2.5.1 removed","accounts_in_message":[],"_revision_number":2},{"id":"bf61c56195a716a106ded9632f87815a47893400","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:06:00.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"7fe9895f11b89448dc19f5785909b7765e898d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 15:14:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34ecfb5ff8d44d6950aa726b1df18a6 : TIMED_OUT in 32m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8cfa16bd4654576adee1f499895173b : TIMED_OUT in 1h 01m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/425fbc51481848e09f7520bae3c09398 : TIMED_OUT in 1h 02m 32s","accounts_in_message":[],"_revision_number":2},{"id":"04ae501030d017bc1e63807ef8089507f0995386","author":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"date":"2021-09-16 15:54:45.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for taking care of this.","accounts_in_message":[],"_revision_number":2},{"id":"9a4a13f4b25e34c9c2061cac3c4045698a513be1","author":{"_account_id":30859,"name":"M Camp","email":"mcamp859@hotmail.com","username":"mcamp859"},"date":"2021-09-19 18:57:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1f175a668de8a6065dd101966d90f20c7c9cfe4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-19 19:08:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602412834a2a4171a7270c2a4442f256 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f0adf4892db43218173d3a0f5a5506e : SUCCESS in 9m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/be0195634a0446239cacc88f0f8905fc : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"38095fd272352f97c6b303785aee50c74d8aa2cb","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-20 09:08:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2269444630c40e4a31e8def5a2572d6f0a4a5726","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-20 14:53:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* PATCH version bump looks OK\n* uploaded by release liaison\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":2},{"id":"e942aa3f916048f62582a6f39694dff68390add4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:54:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"926d34560bb21cd0104beae1cbc2ed46c934d45e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 15:07:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/682617528442440a87b81586b1b283c0 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338b5615e6544763ab4ee6ecb1687b81 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a547585e6a424229b6961998598f65fd : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6dbec4ace7b992853b3294d726042093603a13a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 15:07:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e664b71070693d6e7fc1b7abffd1480367685e02","revisions":{"964ebd75b8b1c6008c2a62e1429454b125d708a7":{"kind":"REWORK","_number":1,"created":"2021-09-16 13:59:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/809419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/809419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/809419/1"}}},"commit":{"parents":[{"commit":"bfe220582f0c71e5caad9df8002183fe1d2dbfd0","subject":"Merge \"Proposing Xena RC1 for tacker-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfe220582f0c71e5caad9df8002183fe1d2dbfd0"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-16 13:57:50.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-16 13:59:29.000000000","tz":120},"subject":"release openstackdocstheme 2.3.1","message":"release openstackdocstheme 2.3.1\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-September/024917.html\n\nChange-Id: I56755457df76321a503ae5cc86690215ce4168a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/964ebd75b8b1c6008c2a62e1429454b125d708a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/964ebd75b8b1c6008c2a62e1429454b125d708a7"}]},"branch":"refs/heads/master"},"e664b71070693d6e7fc1b7abffd1480367685e02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-16 14:02:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/19/809419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/809419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/809419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/809419/2"}}},"commit":{"parents":[{"commit":"bfe220582f0c71e5caad9df8002183fe1d2dbfd0","subject":"Merge \"Proposing Xena RC1 for tacker-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfe220582f0c71e5caad9df8002183fe1d2dbfd0"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-16 13:57:50.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-16 14:01:23.000000000","tz":120},"subject":"Release openstackdocstheme 2.3.1","message":"Release openstackdocstheme 2.3.1\n\nThe StarlingX docs site utilizes a theme variant within the\nopenstackdocstheme repo, but consumes packaged releases. A recent\nchange merged to fix an issue with double heading, and the StarlingX\ncommunity would appreciate a new openstackdocstheme release to\nmake use of it.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-September/024917.html\n\nChange-Id: I56755457df76321a503ae5cc86690215ce4168a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e664b71070693d6e7fc1b7abffd1480367685e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e664b71070693d6e7fc1b7abffd1480367685e02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
