)]}'
{"id":"openstack%2Freleases~809892","triplet_id":"openstack%2Freleases~master~I6992b470a19fb4e200322c723c0cce1202b5318d","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I6992b470a19fb4e200322c723c0cce1202b5318d","subject":"Cinder team releases for stable/victoria","status":"MERGED","created":"2021-09-17 18:31:43.000000000","updated":"2021-09-27 09:34:40.000000000","submitted":"2021-09-27 09:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"809892","meta_rev_id":"ab15282f744db977212ea628b5de350b7b3505d0","_number":809892,"virtual_id_number":809892,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 09:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-09-24 18:11:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-09-27 09:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-27 06:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-09-27 09:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-24 18:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 18:33:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 18:43:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-09-17 18:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 19:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-09-20 09:07:55.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-09-20 14:43:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-27 06:18:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-27 09:22:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"5fea5c3ed3730a3a4db5853da66fc9e800680a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-17 18:31:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21711300718c45c4be5233c432c8a22235218170","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-17 18:43:08.000000000","message":"Patch Set 1: Code-Review+1\n\nRelease versions and hashes LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"796d694cdf07e2bcee9101971011c746be30d182","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:45:10.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5a0a62300c4c7fe0e2e74f2cd48b491e3ddf7829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1064a1bda1854ca6903ce0906812e9c8 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f68cefd607c9495a8d4383c11957c9a5 : SUCCESS in 21m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5ffeab3e7ce43a9a05c1a80e6da1449 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"365621ce7e74b904e1573f9df571d8652e0329f2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-20 09:07:55.000000000","message":"Patch Set 1:\n\nSome of the driver changes in Cinder 17.1.1 are a bit on the feature side, so you could consider 17.2.0 instead. That would draw attention to the upgrade note too. The others LGTM","accounts_in_message":[],"_revision_number":1},{"id":"c3da0b413f1afbab4c4e562bbf5bb9cb754cd944","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-20 12:36:21.000000000","message":"Patch Set 1: -Code-Review\n\nI think Thierry has a good point, let\u0027s see what others on the cinder stable-maint team think.","accounts_in_message":[],"_revision_number":1},{"id":"bf8839607b06a92ac03f3eb46a7fb9b837a47e00","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:37:32.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"ec7cb83be265fb151030f064131d63bd489693ae","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-20 13:23:52.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Some of the driver changes in Cinder 17.1.1 are a bit on the feature side, so you could consider 17.2.0 instead. That would draw attention to the upgrade note too. The others LGTM\n\nThanks Thierry for the review.\nI\u0027m not too sure about the driver features part since we allow backporting driver features and might end up with at least one feature every stable release hence minor version bump everytime.\nThe point regarding the upgrade notes makes sense (as also Brian pointed out on IRC) so i will do a minor bump for it.","accounts_in_message":[],"_revision_number":1},{"id":"b09975976ed0d0db92c01db40eee4cb2297db9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-20 13:24:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"718ded9f1d6e51c9d3f18c475c5245652d7dcf4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 13:34:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482dc0fb7f9d4e2d8bc1ce12d9d53caf : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f2108fa775e4de5a9582bc0dd3eed5a : SUCCESS in 9m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9c71f6c796134943ab94d4fe2dcaf032 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"cbdd55863bc53a880f62399d19d09f888a30ab66","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-20 13:45:46.000000000","message":"Patch Set 2: Code-Review+1\n\nAgree with what Rajat said, and thanks to Thierry for pointing out the importance of emphasizing the upgrade note, I think that was worth the minor version bump.  Change LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"636ee4fe01c977a81f566e7513eed26bbb7c91b9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 13:48:49.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"4af5a0fdeca2be35f026596a40162d0acf77d015","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-20 14:43:35.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM:\n* version bumps are as agreed in the comments\n* approved by PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":2},{"id":"0f491802e905020bc47edfd642449f125b71140a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-22 09:47:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0ab4ceb16fd1b1b84fce467ea881354cc9b41729","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-22 09:49:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08acec5a0218598b4b9ddeb1afef102e15584dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:04:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9ba004f9c2418387a7f7629479d2d6 : SUCCESS in 8m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51419ae80af24cab97d74ae191923c12 : SUCCESS in 8m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bbb662e4a599470f83f79c25049957b8 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0698cd27abb9f48bc883b24161d0921f2942abcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-24 18:06:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d381effd53237f0f14f5c6a53fdbb41ca652e954","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-24 18:11:46.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"8b2369b9363419dff0265c2f199281e6f694633b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 18:13:16.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"631f4231964ac682f39f60d221716eba00761bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 18:20:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83eda43219fb4ad88b744aca416f0080 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73551c895eb34a88a59b7ee761e149c2 : SUCCESS in 8m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7696c8de55044fec9283f98caa217ab3 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"1f6a0f02295f555269b12d9159414cbf5f0c22bb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-27 06:18:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a74996ed052ac9ae9eeb608cd38b2cfe44e6eab4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-27 09:22:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e17d360839e655a5afdb5b9f85902f3e8d0263a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:22:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9f403cf32f0972055811cecf3ecbba961cde2dfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:34:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174a32f98d394ade893c5253df890bcf : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf57d4824114f8c850598556bfd36d3 : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ba99289bc5442dca543d609845637cb : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":4},{"id":"ab15282f744db977212ea628b5de350b7b3505d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:34:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"747ca9c0221cb81274cf54e0fab1614d2fe5dcac","revisions":{"243ac653ee108326f58749c2224cb66b3ef9953d":{"kind":"REWORK","_number":1,"created":"2021-09-17 18:31:43.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/92/809892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/809892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/809892/1"}}},"commit":{"parents":[{"commit":"ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227","subject":"Merge \"Proposing Xena RC1 for heat-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 18:29:50.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 18:31:39.000000000","tz":-240},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (no functional changes in branch)\n- python-brick-cinderclient-ext (no functional changes in branch)\n\nChange-Id: I6992b470a19fb4e200322c723c0cce1202b5318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/243ac653ee108326f58749c2224cb66b3ef9953d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/243ac653ee108326f58749c2224cb66b3ef9953d"}]},"branch":"refs/heads/master"},"7e5de0458b1a61b2f93115f8b9da32aa40f4801f":{"kind":"REWORK","_number":2,"created":"2021-09-20 13:24:27.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/92/809892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/809892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/809892/2"}}},"commit":{"parents":[{"commit":"ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227","subject":"Merge \"Proposing Xena RC1 for heat-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 18:29:50.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-20 13:24:11.000000000","tz":0},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (no functional changes in branch)\n- python-brick-cinderclient-ext (no functional changes in branch)\n\nChange-Id: I6992b470a19fb4e200322c723c0cce1202b5318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e5de0458b1a61b2f93115f8b9da32aa40f4801f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e5de0458b1a61b2f93115f8b9da32aa40f4801f"}]},"branch":"refs/heads/master"},"a04419b1a5ccc6f00d4f817d7ecd1cd2bae1cd13":{"kind":"REWORK","_number":3,"created":"2021-09-22 09:47:24.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/92/809892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/809892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/809892/3"}}},"commit":{"parents":[{"commit":"ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227","subject":"Merge \"Proposing Xena RC1 for heat-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 18:29:50.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-22 09:47:16.000000000","tz":0},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (no functional changes in branch)\n- python-brick-cinderclient-ext (no functional changes in branch)\n\nChange-Id: I6992b470a19fb4e200322c723c0cce1202b5318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a04419b1a5ccc6f00d4f817d7ecd1cd2bae1cd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a04419b1a5ccc6f00d4f817d7ecd1cd2bae1cd13"}]},"branch":"refs/heads/master"},"747ca9c0221cb81274cf54e0fab1614d2fe5dcac":{"kind":"REWORK","_number":4,"created":"2021-09-24 18:06:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/92/809892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/809892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/809892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/809892/4"}}},"commit":{"parents":[{"commit":"ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227","subject":"Merge \"Proposing Xena RC1 for heat-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ac7d88b2ae72df0c5c1e7c68892cc03bcdf0c227"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 18:29:50.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-24 18:05:37.000000000","tz":0},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (no functional changes in branch)\n- python-brick-cinderclient-ext (no functional changes in branch)\n\nChange-Id: I6992b470a19fb4e200322c723c0cce1202b5318d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/747ca9c0221cb81274cf54e0fab1614d2fe5dcac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/747ca9c0221cb81274cf54e0fab1614d2fe5dcac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
