)]}'
{"id":"openstack%2Freleases~810548","triplet_id":"openstack%2Freleases~master~I26594ad03988742fd38b77631002bab6dc40021b","project":"openstack/releases","branch":"master","topic":"tacker-rc2","hashtags":[],"change_id":"I26594ad03988742fd38b77631002bab6dc40021b","subject":"Proposing xena RC2 for tacker","status":"MERGED","created":"2021-09-23 06:32:46.000000000","updated":"2021-09-23 12:24:08.000000000","submitted":"2021-09-23 12:24:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810548-tacker-rc2","meta_rev_id":"76bd7c12cb6889e9cb5e252d70d3a20df6251ed6","_number":810548,"virtual_id_number":810548,"owner":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 12:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2021-09-23 12:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-23 11:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2021-09-23 12:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-09-23 10:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 06:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 10:22:59.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-23 10:22:59.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f91f33727b9cdc2bb15ba5201e875fc85ace7159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-23 06:32:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"556a6282ab3aafe9f9b5f7462eb339d107394092","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 06:34:34.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"8f67157cea84b6af47a27157c3db9c6d082b37ee","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-23 06:40:55.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"93f047e80a39402eb40210c0536e44d98554826f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 06:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb683304c514b8fbe2b21d3b2fc4fe3 : SUCCESS in 10m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f813f3f46d94b8781e13d60c2b68b3e : SUCCESS in 6m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2dc41a9a1231495cb1038959f23d93e5 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a33b84e8ac95f2c6d9146ff67325d048f7eaa08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-23 07:23:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a17ccc3d364782bd781bbe88096d4214a2241bd5","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-23 07:24:12.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"b15168486a0f466915f192fb01aed4506fbfb80f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 07:25:13.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"406083509c29b40d138bee150ac7a9985194c51d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 07:38:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b06b892ae84412a8a6067d61bfe8738 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a40691561124c9ab21473bc4599613a : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6932575c89a44a44b52829a0f5ca4d66 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6a62e12636a2e3480f0c319f92de8367412155b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2021-09-23 10:19:49.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a35a4eb09c3970ed1d59e31150d958e93798f987","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:20:59.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"0af3150a1c893460662e2e67b8e442b3f0f1b123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:28:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d63baef5e94aaabca75ad8ee706f4f : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c18319eb46d4f7185eaa1ae094cccc7 : SUCCESS in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8cde65ca72c346d29df52e2915e5c7cd : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"0f18d2ce138612846ca5708126393ed17a89844a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-23 11:46:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"56d236ed25c41727f971d2b5474f4f2d2a338abf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-23 12:13:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM:\n* rc2 release\n* uploaded by PTL\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":3},{"id":"c4712d09909f501decd0f4fd0b415ce39b542b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:13:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b3b522faff6d4671f3ea32ff3b671ec110b05de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:24:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8772cccd57d4de6b84bf59613699109 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e69fb5de2b480ca66543a4ec5cbf5a : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76b008f2f3e04e21899d4119b91da739 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"76bd7c12cb6889e9cb5e252d70d3a20df6251ed6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:24:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aeb9df892e634508de6634f4852fd4186e601de4","revisions":{"2c782dc65e42b5af578111250060f6107a82c617":{"kind":"REWORK","_number":1,"created":"2021-09-23 06:32:46.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/48/810548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/810548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/810548/1"}}},"commit":{"parents":[{"commit":"09970e77d69582a0914bdf7a29eccf8df262e39e","subject":"Merge \"Release ironic-tempest-plugin for xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09970e77d69582a0914bdf7a29eccf8df262e39e"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 06:31:37.000000000","tz":540},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 06:32:43.000000000","tz":540},"subject":"[WIP] Proposing Xena RC2 for tacker","message":"[WIP] Proposing Xena RC2 for tacker\n\n$ git log --oneline --no-merges 6.0.0.0rc1..0df88fb2\n0df88fb2 (HEAD, origin/stable/xena) Fix cnf rollback after instantiation failure\nc40c14af Add missing options about periodic tasks\nd66ecd14 Update zuul environment to support helm chart\n292e05dc Fix get information about multiple VNF instances\n8e97bf23 Support TOSCA route for ChgExternalConnectivity\n23f3cca4 Drop test for lower constraints\nb22db3d8 (gerrit/stable/xena) Update TOX_CONSTRAINTS_FILE for stable/xena\nfc724184 Update .gitreview for stable/xena\n\nSigned-off-by: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: I26594ad03988742fd38b77631002bab6dc40021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c782dc65e42b5af578111250060f6107a82c617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c782dc65e42b5af578111250060f6107a82c617"}]},"branch":"refs/heads/master"},"4c4bb482d106ddef8877e0e7ed76c74ae22f6799":{"kind":"REWORK","_number":2,"created":"2021-09-23 07:23:34.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/48/810548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/810548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/810548/2"}}},"commit":{"parents":[{"commit":"09970e77d69582a0914bdf7a29eccf8df262e39e","subject":"Merge \"Release ironic-tempest-plugin for xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09970e77d69582a0914bdf7a29eccf8df262e39e"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 06:31:37.000000000","tz":540},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 07:20:01.000000000","tz":540},"subject":"[WIP] Proposing Xena RC2 for tacker","message":"[WIP] Proposing Xena RC2 for tacker\n\n$ git log --oneline --no-merges 6.0.0.0rc1..43b19f63\n43b19f63 (HEAD, origin/stable/xena) Support PVs for k8s cluster with MgmtDriver\nc94228a1 Update instructions of installing k8s VIM\n1474cfe7 Private registry for deploying Kubernetes cluster\n0e10c540 Add groups under topology_template in vnf_dict\n17fed565 Sample script for healing k8s with kubespray\n143c04ab Sample script for scaling k8s with kubespray\n9c9e8827 Sample script for deploying k8s with kubespray\ne48234f9 Sample script of pod-affinity in kubernetes cluster\n34eaab61 Add documents for multi-version API support\nfa73e438 Modify processing of _heal_grant and _scale_grant\nd8b2ddbd Multi version API support\nfa5f63d5 Object definitions of multi version API support\n6205b4d5 Fix zuul settings to use latest kuryr-kubernetes\nf77ae4a9 Support Helm chart as interface for Kubernetes VIM\n9037608c Modify Placement Processing to Aspect definition\n487ee151 Fix for Anti Affinity verification Exception\n0133a35c Fix occasional multinode-sol job FT error\n993169b2 Fix status code for invalid filter conditions\n3eefc875 Add missing [oslo_reports] options\n0df88fb2 Fix cnf rollback after instantiation failure\nc40c14af Add missing options about periodic tasks\nd66ecd14 Update zuul environment to support helm chart\n292e05dc Fix get information about multiple VNF instances\n8e97bf23 Support TOSCA route for ChgExternalConnectivity\n23f3cca4 Drop test for lower constraints\nb22db3d8 (gerrit/stable/xena) Update TOX_CONSTRAINTS_FILE for stable/xena\nfc724184 Update .gitreview for stable/xena\n\nSigned-off-by: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: I26594ad03988742fd38b77631002bab6dc40021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4bb482d106ddef8877e0e7ed76c74ae22f6799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4bb482d106ddef8877e0e7ed76c74ae22f6799"}]},"branch":"refs/heads/master"},"aeb9df892e634508de6634f4852fd4186e601de4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-23 10:19:49.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/48/810548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/810548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/810548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/810548/3"}}},"commit":{"parents":[{"commit":"09970e77d69582a0914bdf7a29eccf8df262e39e","subject":"Merge \"Release ironic-tempest-plugin for xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09970e77d69582a0914bdf7a29eccf8df262e39e"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 06:31:37.000000000","tz":540},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2021-09-23 10:17:32.000000000","tz":540},"subject":"Proposing xena RC2 for tacker","message":"Proposing xena RC2 for tacker\n\n$ git log --oneline --no-merges 6.0.0.0rc1..43b19f63\n43b19f63 (HEAD, origin/stable/xena) Support PVs for k8s cluster with MgmtDriver\nc94228a1 Update instructions of installing k8s VIM\n1474cfe7 Private registry for deploying Kubernetes cluster\n0e10c540 Add groups under topology_template in vnf_dict\n17fed565 Sample script for healing k8s with kubespray\n143c04ab Sample script for scaling k8s with kubespray\n9c9e8827 Sample script for deploying k8s with kubespray\ne48234f9 Sample script of pod-affinity in kubernetes cluster\n34eaab61 Add documents for multi-version API support\nfa73e438 Modify processing of _heal_grant and _scale_grant\nd8b2ddbd Multi version API support\nfa5f63d5 Object definitions of multi version API support\n6205b4d5 Fix zuul settings to use latest kuryr-kubernetes\nf77ae4a9 Support Helm chart as interface for Kubernetes VIM\n9037608c Modify Placement Processing to Aspect definition\n487ee151 Fix for Anti Affinity verification Exception\n0133a35c Fix occasional multinode-sol job FT error\n993169b2 Fix status code for invalid filter conditions\n3eefc875 Add missing [oslo_reports] options\n0df88fb2 Fix cnf rollback after instantiation failure\nc40c14af Add missing options about periodic tasks\nd66ecd14 Update zuul environment to support helm chart\n292e05dc Fix get information about multiple VNF instances\n8e97bf23 Support TOSCA route for ChgExternalConnectivity\n23f3cca4 Drop test for lower constraints\nb22db3d8 (gerrit/stable/xena) Update TOX_CONSTRAINTS_FILE for stable/xena\nfc724184 Update .gitreview for stable/xena\n\nSigned-off-by: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: I26594ad03988742fd38b77631002bab6dc40021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aeb9df892e634508de6634f4852fd4186e601de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aeb9df892e634508de6634f4852fd4186e601de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
