)]}'
{"id":"openstack%2Freleases~810925","triplet_id":"openstack%2Freleases~master~I82377f0b1691f017d684bad56b115ebaa77f9034","project":"openstack/releases","branch":"master","topic":"reqs-xena","hashtags":[],"change_id":"I82377f0b1691f017d684bad56b115ebaa77f9034","subject":"branch requirements project for stable/xena","status":"MERGED","created":"2021-09-24 14:36:57.000000000","updated":"2021-10-01 20:30:53.000000000","submitted":"2021-10-01 20:30:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"810925-reqs-xena","meta_rev_id":"8196b1d2825076228506a493e31e047909e3a132","_number":810925,"virtual_id_number":810925,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-01 20:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2021-10-01 20:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2021-10-01 20:14:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 14:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 17:21:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-30 05:32:12.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"74c808d4a374522509070aebbd423587dd69eb7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-24 14:36:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b23d3fc264edb2ac9bc31f1707a34f5f42110a8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 14:40:52.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"bf5703cd8e458b019fa318dc00cc3974bef0095b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 14:54:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f04546723fc4b3a8e7936312ca48b7c : SUCCESS in 8m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f885be93c13401c8749c708ada523b5 : FAILURE in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ebf6c320249446519b603daed85ff0df : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"aeb675b583c959bf3924f8f9a9d870da5239c7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-24 15:09:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69050e989cf8bc7f2daac013542a20bae34ed03e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 15:12:00.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9c04195ed3da9ddbb67201b9f3904b04e5153597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 15:18:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f9688c77e44a47ac177251f820d3e6 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a85d1e453a4c470781834e017f1d215c : SUCCESS in 6m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/266b162fdc804121a4a4956f93b732d6 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e664cc1de03ed55595f6c0676fe10476e6a7d90b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-24 17:21:25.000000000","message":"Patch Set 2: Workflow-1\n\nWill merge it early next week","accounts_in_message":[],"_revision_number":2},{"id":"2ad36f1cae9502d40db6a44686fb15b6e2d2bf52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-29 15:35:24.000000000","message":"Patch Set 2:\n\nhow early?","accounts_in_message":[],"_revision_number":2},{"id":"8be6487b7ce52fac3571278fe0d866d1a771800a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-30 08:43:58.000000000","message":"Patch Set 2:\n\nos-brick need to release a new version to fix a critical bug.\nWithout that fix operators can\u0027t attach NVMe volume with Nova.\n\nhttps://review.opendev.org/c/openstack/os-brick/+/811886\nhttps://launchpad.net/bugs/1945323","accounts_in_message":[],"_revision_number":2},{"id":"d16989f59e34bd414f1950a227697e7e0b0f96e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-01 07:54:52.000000000","message":"Patch Set 2: -Workflow\n\nos-brick is now out [1]. Waiting for the related reqs updated [2].\nThe hash of this patch need to be updated once the reqs update will be done.\n\n[1] https://review.opendev.org/c/openstack/releases/+/812048\n[2] https://review.opendev.org/c/openstack/requirements/+/812102","accounts_in_message":[],"_revision_number":2},{"id":"09eace90c2641f13523e4aa4f9d77215e8b8c859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-01 07:57:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b0055d4ed8725101d10f0617b22b4d95197c9ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-01 07:58:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b4692dd75fe74333dba3bd7e16d8b0bb2869b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 08:07:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4058b2d31e82408fa18f7be4a086d684 : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e516d568c385455dade5ef1380ae2f3b : FAILURE in 6m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7d387e9cc2b84ca19ad8dc9bd5104f4e : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"643c2fa3f023b0799a0e0c9634cd242bb8783e46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-01 18:04:11.000000000","message":"Patch Set 3:\n\nmerged","accounts_in_message":[],"_revision_number":3},{"id":"65a30544d7c6760f1a414e2c2095607a0aceeea3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-01 19:54:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ce98499bdcdb5573aebd0e496d3d44842add72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:04:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acfb1ce6a7c4a83b9dccfa7ce7dfc7b : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04f4dddb89ba4b8a8b70a5cfda6006dc : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/925d254dac78468886361ff0118d96b8 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3},{"id":"0b36494ba591b43bb8406d7a8591493dcaa0c97e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-01 20:14:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nXena is cut, long life to Yoga!","accounts_in_message":[],"_revision_number":3},{"id":"6a96fd4abe4fca1eb54bd4161ef7c6c1240218cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:15:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"274226b38e9a637d720b049433e744e7f3a79b1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:30:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e518223dd29342cda216c0770c354bbb : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff681e18aa94b4394afcf6fe1ed93c2 : SUCCESS in 9m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2a1d75a88334887ab5be0009467898a : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"8196b1d2825076228506a493e31e047909e3a132","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:30:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4f00b86a40ca7b1017549722a3fa1cc287977039","revisions":{"79eddcec4c20b470dbccda8985051ce02bc2b5eb":{"kind":"REWORK","_number":1,"created":"2021-09-24 14:36:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/25/810925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/810925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/810925/1"}}},"commit":{"parents":[{"commit":"f6c12739012bca253492a3845f2ff6959f7b0fe7","subject":"Merge \"Create stable/xena for grenade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6c12739012bca253492a3845f2ff6959f7b0fe7"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-09-24 14:36:03.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-09-24 14:36:03.000000000","tz":-300},"subject":"branch requirements project for stable/xena","message":"branch requirements project for stable/xena\n\nChange-Id: I82377f0b1691f017d684bad56b115ebaa77f9034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79eddcec4c20b470dbccda8985051ce02bc2b5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79eddcec4c20b470dbccda8985051ce02bc2b5eb"}]},"branch":"refs/heads/master"},"eb38aa58d93b53931a2c4b021349da0ae11f0649":{"kind":"REWORK","_number":2,"created":"2021-09-24 15:09:31.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/25/810925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/810925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/810925/2"}}},"commit":{"parents":[{"commit":"f6c12739012bca253492a3845f2ff6959f7b0fe7","subject":"Merge \"Create stable/xena for grenade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6c12739012bca253492a3845f2ff6959f7b0fe7"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-09-24 14:36:03.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-09-24 15:09:22.000000000","tz":-300},"subject":"branch requirements project for stable/xena","message":"branch requirements project for stable/xena\n\nChange-Id: I82377f0b1691f017d684bad56b115ebaa77f9034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb38aa58d93b53931a2c4b021349da0ae11f0649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb38aa58d93b53931a2c4b021349da0ae11f0649"}]},"branch":"refs/heads/master"},"4f00b86a40ca7b1017549722a3fa1cc287977039":{"kind":"REWORK","_number":3,"created":"2021-10-01 07:57:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/25/810925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/810925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/810925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/810925/3"}}},"commit":{"parents":[{"commit":"f6c12739012bca253492a3845f2ff6959f7b0fe7","subject":"Merge \"Create stable/xena for grenade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6c12739012bca253492a3845f2ff6959f7b0fe7"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-09-24 14:36:03.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-01 07:57:36.000000000","tz":120},"subject":"branch requirements project for stable/xena","message":"branch requirements project for stable/xena\n\nChange-Id: I82377f0b1691f017d684bad56b115ebaa77f9034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f00b86a40ca7b1017549722a3fa1cc287977039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f00b86a40ca7b1017549722a3fa1cc287977039"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
