)]}'
{"id":"openstack%2Freleases~811941","triplet_id":"openstack%2Freleases~master~Ia9cde2755dd0d9c80a42475fdb9f849a6549539b","project":"openstack/releases","branch":"master","topic":"jsonchema","hashtags":[],"change_id":"Ia9cde2755dd0d9c80a42475fdb9f849a6549539b","subject":"Constraints jsonschema to version lower than 4.0.0","status":"MERGED","created":"2021-09-30 09:38:24.000000000","updated":"2021-09-30 10:08:01.000000000","submitted":"2021-09-30 10:08:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811941-jsonchema","meta_rev_id":"f509f04ccc1e106643e79ac7059bb20eb030f095","_number":811941,"virtual_id_number":811941,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 10:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-09-30 09:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-09-30 09:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-09-30 09:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 09:43:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-30 09:47:55.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-09-30 09:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b58aca64f5ce424cc843b2e230ef43f22ba02043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-30 09:38:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee5853a5c535be2d04b03adcaababc59bba42e7b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-30 09:43:18.000000000","message":"Patch Set 1: Code-Review+2\n\nToday is Final RC day, so we are quite in a need for a working validator. This will serve until we fix the issue. Thanks Hervé!","accounts_in_message":[],"_revision_number":1},{"id":"4c5766415cb5ad66fca5ee90ac23e71a6d5e8c93","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-30 09:47:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI like my validators working","accounts_in_message":[],"_revision_number":1},{"id":"ed23d6e8dc2c16324b4be1b0abe79b168f6fd4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ca0e163a85462bbd13b9ae951b3bce : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/897de358091b4299b66560fa780b1ed4 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b44c6a695a8d4d0898ae8f1f2bfb7a23 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4bae9844d34a538dd4f795e86eb231 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/411e1daaabe9413f9900ee2f5f52a250 : SUCCESS in 6m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cfea4125732b4a8a8227b5b341cef860 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb0bdea320054323aacb0c96405a3bf7 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a127f03f248f8fc661c052a353abfed4e91875f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:51:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e8b6d9145afada6e70fc869201808425c6b7013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:08:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e401942c5c4466840735a24ca92674 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7da431672af7499aa4496fe41149792f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/730097e6fc534a759d97c170cbb0abfc : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a04f6019362451f96e2c872fc43d445 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be008323578549bcb3fe6dc4845f88a8 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e696ff7d9fa41e3be30e56b2596f87a : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f509f04ccc1e106643e79ac7059bb20eb030f095","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:08:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40b40d6604200e44e16a9ab825104f457fa1fb07","revisions":{"40b40d6604200e44e16a9ab825104f457fa1fb07":{"kind":"REWORK","_number":1,"created":"2021-09-30 09:38:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/811941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/811941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/811941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/811941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/811941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/811941/1"}}},"commit":{"parents":[{"commit":"192fe15bdeb1739507e80442e67e59e64fcf4a92","subject":"Merge \"Puppet OpenStack Xena Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/192fe15bdeb1739507e80442e67e59e64fcf4a92"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-30 09:36:01.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-30 09:38:14.000000000","tz":120},"subject":"Constraints jsonschema to version lower than 4.0.0","message":"Constraints jsonschema to version lower than 4.0.0\n\nThis patch aim to unlock our validation before spending\nmore time to fix the following traceback:\n\n```\n+ yamllint -f parsable -c ./yamllint.yml deliverables/victoria deliverables/wallaby deliverables/xena deliverables/_independent\nvalidate run-test: commands[1] | /home/zuul/src/opendev.org/openstack/releases/tools/tox-log-command.sh check-schema\nsetting PATH\u003d/home/zuul/src/opendev.org/openstack/releases/.tox/validate/bin:/usr/local/sbin:/usr/local/bin:/usr/sbin:/usr/bin:/sbin:/bin:/usr/games:/usr/local/games:/snap/bin\n[1954] /home/zuul/src/opendev.org/openstack/releases$ /home/zuul/src/opendev.org/openstack/releases/tools/tox-log-command.sh check-schema\nLogging check-schema output to /home/zuul/src/opendev.org/openstack/releases/.tox/validate/log/check-schema-results.log\n   INFO: Checking data/series_status.yaml\nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/releases/.tox/validate/bin/check-schema\", line 10, in \u003cmodule\u003e\n    sys.exit(main())\n  File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/cmds/check_schema.py\", line 103, in main\n    errors.extend(\n  File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/cmds/check_schema.py\", line 66, in validate_one_file\n    validator \u003d make_validator_with_date(schema_data)\n  File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/cmds/check_schema.py\", line 58, in make_validator_with_date\n    return jsonschema.validators.extend(\nTypeError: __init__() got an unexpected keyword argument \u0027types\u0027\nERROR: InvocationError for command /home/zuul/src/opendev.org/openstack/releases/tools/tox-log-command.sh check-schema (exited with code 1)\n```\n\nChange-Id: Ia9cde2755dd0d9c80a42475fdb9f849a6549539b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/40b40d6604200e44e16a9ab825104f457fa1fb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/40b40d6604200e44e16a9ab825104f457fa1fb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
