)]}'
{"id":"openstack%2Freleases~812935","triplet_id":"openstack%2Freleases~master~Ifb9d3b7b342d652783108fc006688acb97713c51","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-10-12 12:20:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb9d3b7b342d652783108fc006688acb97713c51","subject":"Add Yoga deliverable files","status":"MERGED","created":"2021-10-07 08:58:59.000000000","updated":"2021-10-12 12:20:46.000000000","submitted":"2021-10-12 12:20:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812935","meta_rev_id":"9f7e46a0208e1d898340321f1f5cd7f41a31a7a9","_number":812935,"virtual_id_number":812935,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 12:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-10-12 12:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-10-11 12:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-10-12 12:01:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 09:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-07 09:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 09:28:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-07 11:23:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-07 11:38:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-10-07 13:17:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"},{"updated":"2021-10-12 12:01:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"2752c99d8e2c9c62f0bf22c6e96af45c52c104f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-07 08:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94bcd434e9c475fc6f1e4b22013b06519c6e9f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 09:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae912d6667e471195a0c2caeb806415 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2fd560cdd1749baa7d4c6a1165f0b0e : SUCCESS in 17m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19c40993648a48439269ec9d83f20c2f : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"673cf2340f2d8efa4752f5b61fdcfb3ebf8eb2e0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 09:28:33.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM.\n\nWe should notice that some trailing projects are not yet released within Xena and are included here, however we did the same with the previous series. We should just be vigilant and ensure that they released with Xena first.\n\nOnly +1 to let\u0027s time to PTL to approve.","accounts_in_message":[],"_revision_number":1},{"id":"41e9910d20513a5876962e1ea248cdac76596c12","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 09:28:52.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/c/openstack/releases/+/786382","accounts_in_message":[],"_revision_number":1},{"id":"0210bbf45a6eb335ac7b01cebefb9292d0c266e0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-07 11:38:41.000000000","message":"Patch Set 1: Code-Review-1\n\nshall we have tap-as-a-service in this list?","accounts_in_message":[],"_revision_number":1},{"id":"bceca6a2829882620d839553670186b11c26a31c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 13:00:40.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e shall we have tap-as-a-service in this list?\n\nIndeed https://opendev.org/openstack/governance/commit/b8d28176698e376358fc5b9bd6eb23bf70e7fa7c","accounts_in_message":[],"_revision_number":1},{"id":"50e8ff011b3b113b705787a9dfc6e59308013af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-07 13:03:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c43650caf0c074d3591cd8822de99035790d02a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 13:05:35.000000000","message":"Patch Set 1: Code-Review-1\n\nAlso we need to add all the venus deliverables:\n\n- venus\n- venus-specs\n- venus-dashboard\n- python-venusclient\n- venus-tempest-plugin\n\nhttps://opendev.org/openstack/governance/commit/e543ed3445591c407570af3460e6d90ebcbccc82","accounts_in_message":[],"_revision_number":1},{"id":"de45319a633ea23fa74f32f2791ca07a271e0665","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 13:06:46.000000000","message":"Patch Set 2: Code-Review-1\n\nAlso we need to add all the venus deliverables:\n \n - venus\n - venus-specs\n - venus-dashboard\n - python-venusclient\n - venus-tempest-plugin\n \nhttps://opendev.org/openstack/governance/commit/e543ed3445591c407570af3460e6d90ebcbccc82","accounts_in_message":[],"_revision_number":2},{"id":"cefe3d4667cf53f5011e50d1de5ed4b1b9291de7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 13:13:54.000000000","message":"Patch Set 2:\n\nElse, aside the venus deliverables, all the other deliverables seems to be there.","accounts_in_message":[],"_revision_number":2},{"id":"691c5d60abf2e9cf635d2d74755d8c1a8312e302","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-07 13:19:56.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Also we need to add all the venus deliverables:\n\u003e  \n\u003e  - venus\n\u003e  - venus-specs\n\u003e  - venus-dashboard\n\u003e  - python-venusclient\n\u003e  - venus-tempest-plugin\n\u003e  \n\u003e https://opendev.org/openstack/governance/commit/e543ed3445591c407570af3460e6d90ebcbccc82\n\nhmmm, but those are not yet under openstack/ namespace. Shouldn\u0027t we wait until those are moved under openstack?","accounts_in_message":[],"_revision_number":2},{"id":"c6ef19ded523244dc903c5ef29893fdefdc062d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 13:24:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a69d82c7ef84faa886fbb79eb4b6135 : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72e6dd32d29b42dc88554089a5d58419 : SUCCESS in 15m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d28065cab1464cb08e1f3307ec77441f : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"dc851ef4e248030271fa31c566b100ae6c439102","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 17:44:06.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e Also we need to add all the venus deliverables:\n\u003e \u003e  \n\u003e \u003e  - venus\n\u003e \u003e  - venus-specs\n\u003e \u003e  - venus-dashboard\n\u003e \u003e  - python-venusclient\n\u003e \u003e  - venus-tempest-plugin\n\u003e \u003e  \n\u003e \u003e https://opendev.org/openstack/governance/commit/e543ed3445591c407570af3460e6d90ebcbccc82\n\u003e \n\u003e hmmm, but those are not yet under openstack/ namespace. Shouldn\u0027t we wait until those are moved under openstack?\n\nIndeed [1]. I was thinking that the switch was done.\n\n[1] https://opendev.org/openstack/venus","accounts_in_message":[],"_revision_number":2},{"id":"b92d584c054cb52659d4fcde46da06f473ca0f7d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-08 06:45:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9db21973c92bd93f33e9a846ebe1fcbac0b5da7b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-11 12:11:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"473ead8c9e5dc98793d88862e61b971febe1eeff","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-12 12:01:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"508a48a82fe4b8d910f9f5f160d5314460783d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:02:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e23ef5341c62269ceb062811a0969a6c3b88189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:20:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d566013c0f0413f99489172510129db : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda0a97a5523472b96fff4fbf60d3647 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46048c5694834f2b9ab1f6183e6dd63f : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9f7e46a0208e1d898340321f1f5cd7f41a31a7a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:20:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33900ec884b1a76363ed48c4a13838a60a2fd1dc","revisions":{"8272790aa260b59c6e09d430c597d448bd2312be":{"kind":"REWORK","_number":1,"created":"2021-10-07 08:58:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/812935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/812935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/812935/1"}}},"commit":{"parents":[{"commit":"a3e86c35b88a636da79ee2d721c56e75df9163a4","subject":"Merge \"Release python-ironicclient 4.6.3 for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3e86c35b88a636da79ee2d721c56e75df9163a4"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-07 08:54:02.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-07 08:57:19.000000000","tz":120},"subject":"Add Yoga deliverable files","message":"Add Yoga deliverable files\n\nThe generated content initializes the Yoga cycle\u0027s deliverable files\nbased on xena deliverables.\n\nChange-Id: Ifb9d3b7b342d652783108fc006688acb97713c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8272790aa260b59c6e09d430c597d448bd2312be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8272790aa260b59c6e09d430c597d448bd2312be"}]},"branch":"refs/heads/master"},"33900ec884b1a76363ed48c4a13838a60a2fd1dc":{"kind":"REWORK","_number":2,"created":"2021-10-07 13:03:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/812935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/812935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/812935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/812935/2"}}},"commit":{"parents":[{"commit":"a3e86c35b88a636da79ee2d721c56e75df9163a4","subject":"Merge \"Release python-ironicclient 4.6.3 for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3e86c35b88a636da79ee2d721c56e75df9163a4"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-07 08:54:02.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-07 13:03:26.000000000","tz":120},"subject":"Add Yoga deliverable files","message":"Add Yoga deliverable files\n\nThe generated content initializes the Yoga cycle\u0027s deliverable files\nbased on xena deliverables.\n\nChange-Id: Ifb9d3b7b342d652783108fc006688acb97713c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33900ec884b1a76363ed48c4a13838a60a2fd1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33900ec884b1a76363ed48c4a13838a60a2fd1dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
