)]}'
{"id":"openstack%2Freleases~813201","triplet_id":"openstack%2Freleases~master~I6ca3b79e70b409ad92ae1e19c0aee7eed08c5aa8","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I6ca3b79e70b409ad92ae1e19c0aee7eed08c5aa8","subject":"WIP: Add Extended Maintenance transition steps to process","status":"ABANDONED","created":"2021-10-08 15:07:37.000000000","updated":"2023-10-27 14:56:34.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9951b629b77dee7f9f8040723b84d5b81875b42b","_number":813201,"virtual_id_number":813201,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-08 16:55:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 15:32:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-10-08 16:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"326f0de6f9f0d8154bb8eb02da20be4c8fa04d5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-08 15:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a3cdad1130d816854bf0f5142a7ee45ddf473c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-08 15:32:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c2c2820fdaa96d224724b69f48dc4b75d3bbc30","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-08 15:56:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f3c53ed80b6334a12c2013a8e754c739edb4dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-08 16:03:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b4fbfaa36325de32c2192a5fc747a3445fc3f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 16:17:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e9d43696ea48ec85aaf0d4aab4eef2 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6b44bc35f534631ba46e686e9620a2adb624e00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-08 16:39:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa15009f6098b6615dd8a3164a04297d7963eeaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 16:55:37.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5e0b84905c46b0b6010cfdb19a3bb8 : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9951b629b77dee7f9f8040723b84d5b81875b42b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-10-27 14:56:34.000000000","message":"Abandoned\n\nExtended Maintenance is replaced by the Unmaintained phase (see resolutin in TC repo). Abandoning.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e6d2bda7d6bef82fba68a3427599685db8784c3","revisions":{"32a7b9aad4f13897eb6fbb433584867bf51e88bb":{"kind":"REWORK","_number":1,"created":"2021-10-08 15:07:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/01/813201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/813201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/813201/1"}}},"commit":{"parents":[{"commit":"d071580e2b2183c26fa2345aad44c4f8d1524bb2","subject":"Merge \"Release reno 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d071580e2b2183c26fa2345aad44c4f8d1524bb2"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 14:43:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 15:06:53.000000000","tz":120},"subject":"WIP: Add Extended Maintenance transition steps to process","message":"WIP: Add Extended Maintenance transition steps to process\n\nThe transition to Extended Maintenance of the last but three stable\nseries [1] should happen during a less busy period, practically, should\nbe prepared after the final release of the last series.\n\nAfter the transition it is not possible to release from that stable\nseries anymore. Because of this, projects should decide whether to\nprepare a final release for the given series.\n\nThe final release should be prepared well before the exact transition\ndate to avoid broken releases that cannot be fixed anymore.\n\nThis is why the process is extended with the following steps:\n- at the week after the release: remind the teams to do final releases\n- 1 month after the release: do the actual transition\n\n[1] for example:\n- the current series, which is under development, is Yoga (future\n  delivery)\n- the recently released stable series is Xena\n- the \u0027last but three stable series\u0027 is Ussuri\n\nChange-Id: I6ca3b79e70b409ad92ae1e19c0aee7eed08c5aa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32a7b9aad4f13897eb6fbb433584867bf51e88bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32a7b9aad4f13897eb6fbb433584867bf51e88bb"}]},"branch":"refs/heads/master"},"cf96ed2e825baf608ecac5afa335e349ce81d623":{"kind":"REWORK","_number":2,"created":"2021-10-08 16:03:38.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/01/813201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/813201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/813201/2"}}},"commit":{"parents":[{"commit":"d071580e2b2183c26fa2345aad44c4f8d1524bb2","subject":"Merge \"Release reno 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d071580e2b2183c26fa2345aad44c4f8d1524bb2"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 14:43:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 16:00:55.000000000","tz":120},"subject":"WIP: Add Extended Maintenance transition steps to process","message":"WIP: Add Extended Maintenance transition steps to process\n\nThe transition to Extended Maintenance of the last but three stable\nseries [1] should happen during a less busy period, practically, should\nbe prepared after the final release of the last series.\n\nAfter the transition it is not possible to release from that stable\nseries anymore. Because of this, projects should decide whether to\nprepare a final release for the given series.\n\nThe final release should be prepared well before the exact transition\ndate to avoid broken releases that cannot be fixed anymore.\n\nThis is why the process is extended with the following steps:\n- at the week after the release: remind the teams to do final releases\n- 1 month after the release: do the actual transition\n- between milestone-2 and milestone-3 propose convenience releases for\n  the next EM series to avoid rush and broken releases at EM transition\n\n[1] for example:\n- the current series, which is under development, is Yoga (future\n  delivery)\n- the recently released stable series is Xena\n- the \u0027last but three stable series\u0027 is Ussuri\n\nChange-Id: I6ca3b79e70b409ad92ae1e19c0aee7eed08c5aa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf96ed2e825baf608ecac5afa335e349ce81d623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf96ed2e825baf608ecac5afa335e349ce81d623"}]},"branch":"refs/heads/master"},"7e6d2bda7d6bef82fba68a3427599685db8784c3":{"kind":"REWORK","_number":3,"created":"2021-10-08 16:39:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/01/813201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/813201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/813201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/813201/3"}}},"commit":{"parents":[{"commit":"d071580e2b2183c26fa2345aad44c4f8d1524bb2","subject":"Merge \"Release reno 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d071580e2b2183c26fa2345aad44c4f8d1524bb2"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 14:43:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-10-08 16:39:30.000000000","tz":120},"subject":"WIP: Add Extended Maintenance transition steps to process","message":"WIP: Add Extended Maintenance transition steps to process\n\nThe transition to Extended Maintenance of the last but three stable\nseries [1] should happen during a less busy period, practically, should\nbe prepared after the final release of the last series.\n\nAfter the transition it is not possible to release from that stable\nseries anymore. Because of this, projects should decide whether to\nprepare a final release for the given series.\n\nThe final release should be prepared well before the exact transition\ndate to avoid broken releases that cannot be fixed anymore.\n\nThis is why the process is extended with the following steps:\n- at the week after the release: remind the teams to do final releases\n- 1 month after the release: do the actual transition\n- between milestone-2 and milestone-3 propose convenience releases for\n  the next EM series to avoid rush and broken releases at EM transition\n\n[1] for example:\n- the current series, which is under development, is Yoga (future\n  delivery)\n- the recently released stable series is Xena\n- the \u0027last but three stable series\u0027 is Ussuri\n\nChange-Id: I6ca3b79e70b409ad92ae1e19c0aee7eed08c5aa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e6d2bda7d6bef82fba68a3427599685db8784c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e6d2bda7d6bef82fba68a3427599685db8784c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
