)]}'
{"id":"openstack%2Freleases~813852","triplet_id":"openstack%2Freleases~master~I302621c1b61260c99b91ea0522997619418baedd","project":"openstack/releases","branch":"master","topic":"tripleo-xena","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2021-10-18 16:33:20.000000000","reason":"Change was submitted"},"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2021-10-18 16:33:20.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2021-10-13 19:54:33.000000000","reason":"David Peacock replied on the change"},"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2021-10-13 19:29:25.000000000","reason":"Gael Chamoulaud replied on the change"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2021-10-13 16:37:50.000000000","reason":"mbu replied on the change"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-10-18 09:11:59.000000000","reason":"Thierry Carrez replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-10-18 16:15:55.000000000","reason":"Elod Illes replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-10-18 09:11:06.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2021-10-15 07:35:33.000000000","reason":"Alfredo Moralejo replied on the change"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2021-10-14 14:36:29.000000000","reason":"James Slagle replied on the change"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2021-10-18 16:33:20.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-10-18 14:52:41.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I302621c1b61260c99b91ea0522997619418baedd","subject":"Create releases for tripleo releated projects","status":"MERGED","created":"2021-10-13 15:23:07.000000000","updated":"2021-10-18 16:33:20.000000000","submitted":"2021-10-18 16:33:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"813852-tripleo-xena","meta_rev_id":"887636cd60f413d662276a4b7d044573f7360689","_number":813852,"virtual_id_number":813852,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 16:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2021-10-18 09:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-10-18 09:11:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-10-18 16:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-10-14 14:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2021-10-15 07:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2021-10-18 14:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-18 09:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-10-18 16:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-10-14 14:40:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 15:23:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-10-13 15:23:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-10-13 15:23:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-10-13 15:23:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-10-13 15:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 15:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 16:10:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-10-13 16:10:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-10-13 16:10:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-10-18 09:11:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-10-18 09:11:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-10-18 09:11:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-10-18 09:11:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-18 09:12:15.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"7d2fd7936c87945b0998067e0c04bb4c11ca3f10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-13 15:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3ad9a248994c3c6b326b1b4cb8fb4db8d17614d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-13 15:24:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aefc3b780dfe29e2f377bc3114fe0f2333907b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:35:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b260fb2a744555945a1fec9060fc1f : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6cbe0d1e2a2450cb83deb8f5592e04f : FAILURE in 8m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b61354dc20af4aff9dfa932a27d73144 : FAILURE in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2dad5dc39feb1d34d941ffd6933d979a362dd25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-13 15:45:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7ddb195e6f3f80f4a04dc6ccd13425c765d492b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:01:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd5100f07e9423689faca2cfe64358d : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea5b4420566b47eba54737235d15c827 : FAILURE in 10m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bf0126d4bbe946edbd349d24e87d5cc9 : FAILURE in 5m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c6a8570c157b8fbee8246db4b8d392b0839d81b4","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-10-13 16:37:50.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34fb0d00f8abdbfdca8ceb41f0a3174333ee69a8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-10-13 19:29:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"75768667564f8dd1d9663de9de25312d3743fa37","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-10-13 19:54:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2ff6d3a873a63e8bb9781cc94ac6bb9eaab871ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-13 21:03:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"feb12527905f92b32a89ee25a7da0c4a7020dec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:14:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5260c5ef5a2a4bbc8ac1ed99989e6831 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc07bb4c40844caa81875d71e682b4ab : FAILURE in 9m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/14ab24f33a4e40629189cf982cc042ca : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":3},{"id":"1ed9966286a621a4d07dd4ef483679a5f34e473a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-13 21:57:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"80dfa1a7d065713558f57700b3c45c72faed1997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 22:10:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3f30ab335145cb8e59324c1a12b61a : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aa2f1c77f2c4ca3ab3cb6e2fe03029e : FAILURE in 7m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/366c717fe86241c98d3d84f6882207e5 : FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":4},{"id":"778524566cce2ede0ceb7c41e48ef9f437295368","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-14 13:16:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19819b14a96500ec6865227a1269748d095538e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 13:29:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6886ad3935cf4bd6b89e95eeb2536d78 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2aab15407d5b4016b022cc58399df6cf : SUCCESS in 7m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7f29f9302564916bc95cb8f12ddb5c9 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":4},{"id":"cf4a1d8463886e52fdf26e82efcbfc0a6cc78a77","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-10-14 14:36:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bdad640588e0157685f70b3162f236df62078b17","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 14:40:48.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"a41ec53e2aa3e0cae4883d4ffad9f7fe2031300b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-15 07:35:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3e83fdce2d0d65044bd9d5c29dbce8bc593a20da","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-15 07:44:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7132fe3a60572bcc81eae99fcd20cf7336dfa36d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-15 13:50:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f31e968dad7b203d257fea8bfc027fc59bc5ac57","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-18 09:11:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"442d109c85d4232819eef03e2f3b35b42f3e8c87","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-18 09:11:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"07fcdff6f52d6294f06e208b49cf71b72f3195fe","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-10-18 09:12:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"93d68f825f57027858b177cdc7fad4e5b8f5742c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-18 14:52:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"69c39e83532533986fc505166f58568bd38cd146","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-18 16:15:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c14b0f73adeaf2712a8faf4e0e791c6e02a120dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 16:16:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"caf2dc1aacf853797bd6a469f0cfae12ff1f7ad4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 16:33:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a69e03298e64306b77572d29abe982f : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c176376cfc4789b3703c42a8739657 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2456e81807345feb43b23dd9610f4fe : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":4},{"id":"887636cd60f413d662276a4b7d044573f7360689","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 16:33:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"835a2698645c11239534ad752077f17f784e6fb5","revisions":{"2e33d7ae70efb7a2f0f9f72a2e4c94146d0ca80e":{"kind":"REWORK","_number":1,"created":"2021-10-13 15:23:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/52/813852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/813852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/813852/1"}}},"commit":{"parents":[{"commit":"97528f52185c219fc03389aa66f6133466e21d08","subject":"Merge \"Update release and branch to Wallaby for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97528f52185c219fc03389aa66f6133466e21d08"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:21:40.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:21:40.000000000","tz":-360},"subject":"[WIP] Create releases for tripleo releated projects","message":"[WIP] Create releases for tripleo releated projects\n\nTo simplify RDO releasing Xena, we would like to to cut tags for\nthe current tripleo state.\n\nTODO: need to update hashes with last promoted content after\npuppet-tripleo tag lands\n\nChange-Id: I302621c1b61260c99b91ea0522997619418baedd\nDepends-On: https://review.opendev.org/c/openstack/puppet-tripleo/+/813847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e33d7ae70efb7a2f0f9f72a2e4c94146d0ca80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e33d7ae70efb7a2f0f9f72a2e4c94146d0ca80e"}]},"branch":"refs/heads/master"},"0fa306b1561597ef0a834ab2731852e4e99db3e6":{"kind":"REWORK","_number":2,"created":"2021-10-13 15:24:21.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/52/813852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/813852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/813852/2"}}},"commit":{"parents":[{"commit":"97528f52185c219fc03389aa66f6133466e21d08","subject":"Merge \"Update release and branch to Wallaby for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97528f52185c219fc03389aa66f6133466e21d08"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:21:40.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:24:17.000000000","tz":-360},"subject":"[WIP] Create releases for tripleo releated projects","message":"[WIP] Create releases for tripleo releated projects\n\nTo simplify RDO releasing Xena, we would like to to cut tags for\nthe current tripleo state.\n\nTODO: need to update hashes with last promoted content after\npuppet-tripleo tag lands\n\nChange-Id: I302621c1b61260c99b91ea0522997619418baedd\nDepends-On: https://review.opendev.org/c/openstack/puppet-tripleo/+/813847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0fa306b1561597ef0a834ab2731852e4e99db3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0fa306b1561597ef0a834ab2731852e4e99db3e6"}]},"branch":"refs/heads/master"},"1792483487ef11f992259e37b45f68318e8d22e7":{"kind":"REWORK","_number":3,"created":"2021-10-13 15:45:56.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/52/813852/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/813852/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/813852/3"}}},"commit":{"parents":[{"commit":"97528f52185c219fc03389aa66f6133466e21d08","subject":"Merge \"Update release and branch to Wallaby for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97528f52185c219fc03389aa66f6133466e21d08"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:21:40.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:45:07.000000000","tz":-360},"subject":"[WIP] Create releases for tripleo releated projects","message":"[WIP] Create releases for tripleo releated projects\n\nTo simplify RDO releasing Xena, we would like to to cut tags for\nthe current tripleo state.\n\nTODO: need to update hashes with last promoted content after\npuppet-tripleo tag lands\n\nChange-Id: I302621c1b61260c99b91ea0522997619418baedd\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/813854\nDepends-On: https://review.opendev.org/c/openstack/puppet-tripleo/+/813847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1792483487ef11f992259e37b45f68318e8d22e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1792483487ef11f992259e37b45f68318e8d22e7"}]},"branch":"refs/heads/master"},"835a2698645c11239534ad752077f17f784e6fb5":{"kind":"REWORK","_number":4,"created":"2021-10-13 21:57:35.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/52/813852/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/813852/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/813852/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/813852/4"}}},"commit":{"parents":[{"commit":"97528f52185c219fc03389aa66f6133466e21d08","subject":"Merge \"Update release and branch to Wallaby for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97528f52185c219fc03389aa66f6133466e21d08"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 15:21:40.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-10-13 21:57:17.000000000","tz":-360},"subject":"Create releases for tripleo releated projects","message":"Create releases for tripleo releated projects\n\nTo simplify RDO releasing Xena, we would like to to cut tags for\nthe current tripleo state.\n\nChange-Id: I302621c1b61260c99b91ea0522997619418baedd\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/813854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/835a2698645c11239534ad752077f17f784e6fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/835a2698645c11239534ad752077f17f784e6fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
