)]}'
{"id":"openstack%2Freleases~815334","triplet_id":"openstack%2Freleases~master~Ie0e7c22fff759016b10b3f4c143928323dfb90b4","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2021-10-27 11:29:19.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-10-27 11:29:19.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-10-27 11:16:53.000000000","reason":"Elod Illes replied on the change"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2021-10-27 11:29:19.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-10-26 12:48:02.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"Ie0e7c22fff759016b10b3f4c143928323dfb90b4","subject":"Release horizon 20.2.0(Yoga)","status":"MERGED","created":"2021-10-25 16:19:00.000000000","updated":"2021-10-27 11:29:19.000000000","submitted":"2021-10-27 11:29:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"815334","meta_rev_id":"035ea5d33fbb7ef4ec8a977f3b24af5fc13400f5","_number":815334,"virtual_id_number":815334,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 11:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-10-27 11:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-10-27 10:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2021-10-27 11:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-10-27 09:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 16:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 19:54:42.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-25 19:54:42.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-10-26 12:09:18.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-10-26 17:55:52.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"9227dc84edde73ad460e7628dbbfa6deea0eda82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 16:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"240d0ca35fdc72496c849a4237a1b373d58b3e3a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 16:19:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"7e720d55c1338b0d4d74364b26d896db2b8d43ff","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:20:20.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"be889dacc42be461bae98270b3d106c698b9fb2c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 16:26:41.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":1},{"id":"719b8db58b8fcc2a4955674645863e85c535eeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:27:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6632a74f0c024ff5b29a029bf27a88a5 : FAILURE in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caf368cc7775412d81999ba556525f38 : FAILURE in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b16008d8652a480ea7d989e11b32fb37 : FAILURE in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"505af066bda5168750405fd67c665b3730e6bf26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 16:32:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0871896a9736b025a8b31c7180474c391b4de256","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:34:15.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"82c672dfa6a889093047321a2eea5eca361f30bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:44:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37eaece0b794761af6a0a5494728d5d : FAILURE in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cc6bcef7b5e426994cde9279d60ff28 : FAILURE in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19f5d87ab10d433fadc7c1f2c72cf95b : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5d188f4585573cc5eef1ffa57f38d1c682b855b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 17:12:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eecb8b5af3687c71b45553fb49979d93ebd965ee","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:13:30.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"0d68309b13862557a8c21ea11e36b7a58237f6d2","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 17:13:53.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":3},{"id":"c34833e5d1fa325b2c9d57ed9b155f0afd37bf2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:38:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d8be50eb224c19bdee7f914decd09b : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1781cc8d700d4d90b236bd5afddb6d3d : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/efa9a82665f14f47a9ad87e73fbee86b : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"b36c874a9644c497466eaceaf0edaa46d76c2fa1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 19:53:16.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3a0d546c3f514645b00adcc267e8e57bb4b9a8f4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-25 19:53:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"6c12975cfc2ccee7990582159f002ecf9ad04be9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 19:56:36.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"42eefef6e459b87ac7d4b21b5c74c9964cd16a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 20:04:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6967ab5570fd4c1e82fd381e9bea24bf : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53efa713606d4f988d5f3e869901f54e : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d26aa9f86adc45958cf8bbf8c2a00640 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"4cfb59d7b3b54a526b3474526cd4674965226396","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 04:07:26.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"793596d1ed265a5b72efc4a973a2661dc0d2e9c6","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 04:07:29.000000000","message":"Removed Workflow-1 by Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"a273f37e2bf2361dfff8145bdba1df435acd96d1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 04:07:36.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"f327bee011b7fcf1ddfea63b3b3ab0b6204bb4d4","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 04:07:38.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":4},{"id":"de60f8281ef099021819c673e2a5b0f5fe56b5e1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 04:08:48.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"db20f241eb935b067e3ca95298b7bcad08923d1b","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 04:17:34.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":4},{"id":"56ed6479ad86a24f4897b97234935fd99ceef784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-26 12:09:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4fd0ac6c1d9f9d3b24ca91fe14837e58669270ba","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 12:10:57.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"f9ea3373c8cf792ce3d465f5ba3950be2bd1233f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 12:21:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e46ede31ac44b697d969b5e1784113 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/140c404a1a2f43a39e4d16a0c0c38e50 : SUCCESS in 7m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dea786a1728f4da3b7f75067b51a03a4 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c977789f5c3e97c41fe085d1224d51964fa187d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-26 12:48:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec018d48efdf451ce452a38ed8cd3c1b6d8fb075","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-10-26 13:30:54.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d76f8115e5c705814c284b1c30b0d5228dfe426b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-10-27 07:10:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a5bce4f9677462a38d159746de07c04abc9671a0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-27 07:46:43.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"647d57c8e909e7cbc405ecfc2a7e6d9f6e2c1c32","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-27 07:50:42.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"a3c10857e544f20159499d5ecdbbff06cebfae55","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-27 09:21:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7cc117fe55ef866de51178fb66b58e1305bb4915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-27 09:50:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dce8fbf743da562106448093041afa52ee674af0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-27 09:51:34.000000000","message":"Patch Set 5: -Workflow\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"839abd63f24b40679c0ad446a8bd85513e6bd47a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 09:52:16.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"b1899c793edd5c43040d09779427afdca7f83afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 10:08:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bd3274011146d2809f8da972568a74 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62b0ef3825bc4516a6412fb8bd5e0d3e : SUCCESS in 6m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b04fe8bdfca0451f919d33e935216917 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":6},{"id":"ef48c957ac283954213910fa034f98006e74f41e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-27 10:16:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5876c6c67fbabb78ee21d487f313b26600042a90","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-27 11:16:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"44be6a08da3304bff73bcb911d000a8703de43fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:17:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"877fbc4188749b72ad305b74bd9fbcf433d4a4b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:29:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5aa0d3c3b2432d9368b963e4d14712 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93783c14d21d493089311e650f26f00e : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c70788e755c4f1fab75e042b6560cf0 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":6},{"id":"035ea5d33fbb7ef4ec8a977f3b24af5fc13400f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:29:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"710905272533be4258f600d30b3d87b91e0a72d1","revisions":{"680d3a0e79e33ed19fb2a6e1c231db02ee26614d":{"kind":"REWORK","_number":1,"created":"2021-10-25 16:19:00.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/1"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"subject":"[DNM] Release horizon 21.0.0(Yoga)","message":"[DNM] Release horizon 21.0.0(Yoga)\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/680d3a0e79e33ed19fb2a6e1c231db02ee26614d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/680d3a0e79e33ed19fb2a6e1c231db02ee26614d"}]},"branch":"refs/heads/master"},"febcdd53bcfee13232839987e09008b4adf2d0e3":{"kind":"REWORK","_number":2,"created":"2021-10-25 16:32:33.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/2"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:32:01.000000000","tz":330},"subject":"[DNM] Release horizon 21.0.0(Yoga)","message":"[DNM] Release horizon 21.0.0(Yoga)\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/febcdd53bcfee13232839987e09008b4adf2d0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/febcdd53bcfee13232839987e09008b4adf2d0e3"}]},"branch":"refs/heads/master"},"f0e278c4984028e20a5cf88fb80635da825da384":{"kind":"REWORK","_number":3,"created":"2021-10-25 17:12:08.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/3"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 17:11:49.000000000","tz":330},"subject":"[DNM] Release horizon 21.0.0(Yoga)","message":"[DNM] Release horizon 21.0.0(Yoga)\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e278c4984028e20a5cf88fb80635da825da384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e278c4984028e20a5cf88fb80635da825da384"}]},"branch":"refs/heads/master"},"1a4a37aa77cd717ad8e7edd8982828f09bac17f8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-25 19:53:16.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/4"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 19:41:50.000000000","tz":330},"subject":"Release horizon 21.0.0(Yoga)","message":"Release horizon 21.0.0(Yoga)\n\nThis is the first release for Yoga. horizon default Django version\nis bumped to Django 3.2 from 2.2 [1]. So this patch cut a new release\nso that openstack/requirements picks new horizon version with these\nchanges.\n\n[1] https://review.opendev.org/c/openstack/horizon/+/811412\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a4a37aa77cd717ad8e7edd8982828f09bac17f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a4a37aa77cd717ad8e7edd8982828f09bac17f8"}]},"branch":"refs/heads/master"},"5fab31314c711449ceb1de1271acf156d028427b":{"kind":"REWORK","_number":5,"created":"2021-10-26 12:09:03.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/5"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-26 12:07:15.000000000","tz":330},"subject":"Release horizon 21.0.0(Yoga)","message":"Release horizon 21.0.0(Yoga)\n\nhorizon support Django 3.2 and Django version bump to 3.2\nin requirements.txt in the horizon. So this patch cut a new release\nof the horizon so that all plugins can consume this change.\nFor more information please see [1]\n\n[1] https://review.opendev.org/c/openstack/horizon/+/815393\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5fab31314c711449ceb1de1271acf156d028427b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5fab31314c711449ceb1de1271acf156d028427b"}]},"branch":"refs/heads/master"},"710905272533be4258f600d30b3d87b91e0a72d1":{"kind":"REWORK","_number":6,"created":"2021-10-27 09:50:43.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/815334/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/815334/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/815334/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/815334/6"}}},"commit":{"parents":[{"commit":"4d4ed2ba5676c24b84743ef5f1ac09de879aab21","subject":"Merge \"release oslo.context for xena and wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d4ed2ba5676c24b84743ef5f1ac09de879aab21"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-25 16:17:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-27 09:49:38.000000000","tz":330},"subject":"Release horizon 20.2.0(Yoga)","message":"Release horizon 20.2.0(Yoga)\n\nThis is the first release for Yoga. The minor version is bumped as we\nhave changes in requirements.txt(Django\u003e\u003d2.2,\u003c3.0 changed to \u003e\u003d2.2,\u003c3.3).\nHorizon bumps the upper version of Django to Django3.2. So this patch\ncut a new release of the horizon so that all plugins can consume\nthis change. For more information please see [1]\n\n[1] https://review.opendev.org/c/openstack/horizon/+/815393\n\nChange-Id: Ie0e7c22fff759016b10b3f4c143928323dfb90b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/710905272533be4258f600d30b3d87b91e0a72d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/710905272533be4258f600d30b3d87b91e0a72d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
