)]}'
{"id":"openstack%2Freleases~816557","triplet_id":"openstack%2Freleases~master~If9fb38c926821ab1a99ae5c6eb2f1503ac34ab06","project":"openstack/releases","branch":"master","topic":"manila","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2021-11-15 12:00:24.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-11-15 12:00:24.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-11-13 09:15:26.000000000","reason":"Elod Illes replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-11-12 14:50:38.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"If9fb38c926821ab1a99ae5c6eb2f1503ac34ab06","subject":"[manila] Release manila 10.2.0","status":"MERGED","created":"2021-11-03 20:10:27.000000000","updated":"2021-11-15 12:00:24.000000000","submitted":"2021-11-15 12:00:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816557-manila","meta_rev_id":"f87e3b2223deadb88bd420b74c84690552c21f81","_number":816557,"virtual_id_number":816557,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 12:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-11-09 16:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-11-13 09:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-11-15 11:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-11-15 11:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-11-09 16:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 20:11:41.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-11-03 20:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-03 20:22:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 16:57:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-09 16:57:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"07836430b624b366065cbc2312dc49c6804d2493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-03 20:10:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0070126f18fab339c4397d299819a7a2170d7ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:22:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7458520750584b23b70f8e28529f89c4 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f185270c9494dcbb5c00fa232b39ae6 : SUCCESS in 10m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e26aa231cb6f4daa86064e29118584c3 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f43530c38305e86676115431919b3556148c8932","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-05 21:31:29.000000000","message":"Topic set to manila","accounts_in_message":[],"_revision_number":1},{"id":"0ed746f63e89f88935798acca6824016196fa3a5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-05 21:39:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b6cd9114cb8fbe3fa800e2aed0c15779be7e151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-08 22:10:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fe9446ef08ca26d4402c817b05035e99ffdd8d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-08 22:11:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"74df637120d318c54bdebd7b418616f226bd212d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-08 22:11:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"298b2d7a6011a354e6c0cc3357971313a79072a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:22:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52884ae665744a3393abbd83bd5d16b7 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f34be34f1f048588b1ccda891356c2f : SUCCESS in 6m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0d30414c1b504dfea8cecc7703e4440b : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":3},{"id":"bce07f2f2f9f53e1b4631e5c46465c7f481dedab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-09 16:57:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe0a76705fc4ce4972699aae0d6bdee29cad791c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 16:59:01.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"7014e209138dbd3c915d50852e22e40786db7e2b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-12 11:56:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"beeaea436bf1bac616bca3432a889dafcb24f6f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-12 14:50:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4541fd534a7d22af7c8c769e8edc42ddc7576b59","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-12 21:31:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"491909adce6762c2029095f381d0421939ffbff1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-13 09:15:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"243cb04c94cd66f925e68eff9a54d6f662273e65","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-15 11:47:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99903f5124ce42b919e878fa0d9e8cabbf0dac13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 11:48:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e233fba403581c9bd372e157a208527c6c1b8744","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 12:00:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d4e9e250d74f8fbb9f3bd254ee4c78 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6211b34e8adc4fa3aeeed577f0aed6b5 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bffdfcf7e04847aca794da9bebb07fb9 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f87e3b2223deadb88bd420b74c84690552c21f81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 12:00:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84085d99b69af8aefd363dba30585e642a4cb08e","revisions":{"f6abfb65b87cf2ca10ce097c4eba2552b6246bca":{"kind":"REWORK","_number":1,"created":"2021-11-03 20:10:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/57/816557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/816557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/816557/1"}}},"commit":{"parents":[{"commit":"b78346c06689ecfe14c200dd3196bfbfd158a067","subject":"Release tripleo-operator-ansible 0.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78346c06689ecfe14c200dd3196bfbfd158a067"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-03 20:09:15.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-03 20:10:12.000000000","tz":-180},"subject":"[manila] Release manila 10.1.1","message":"[manila] Release manila 10.1.1\n\nThis release includes several bugfixes and enhancements committed\nagainst the stable/ussuri branch.\n\nChange-Id: If9fb38c926821ab1a99ae5c6eb2f1503ac34ab06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6abfb65b87cf2ca10ce097c4eba2552b6246bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6abfb65b87cf2ca10ce097c4eba2552b6246bca"}]},"branch":"refs/heads/master"},"b3ae6261d11c016a1bc6dad8db12c7b078b28c54":{"kind":"REWORK","_number":2,"created":"2021-11-08 22:10:50.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/57/816557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/816557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/816557/2"}}},"commit":{"parents":[{"commit":"b78346c06689ecfe14c200dd3196bfbfd158a067","subject":"Release tripleo-operator-ansible 0.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78346c06689ecfe14c200dd3196bfbfd158a067"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-03 20:09:15.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-11-08 22:10:37.000000000","tz":0},"subject":"[manila] Release manila 10.1.1","message":"[manila] Release manila 10.1.1\n\nThis release includes several bugfixes and enhancements committed\nagainst the stable/ussuri branch.\n\nChange-Id: If9fb38c926821ab1a99ae5c6eb2f1503ac34ab06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3ae6261d11c016a1bc6dad8db12c7b078b28c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3ae6261d11c016a1bc6dad8db12c7b078b28c54"}]},"branch":"refs/heads/master"},"84085d99b69af8aefd363dba30585e642a4cb08e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-08 22:11:01.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/57/816557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/816557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/816557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/816557/3"}}},"commit":{"parents":[{"commit":"b78346c06689ecfe14c200dd3196bfbfd158a067","subject":"Release tripleo-operator-ansible 0.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78346c06689ecfe14c200dd3196bfbfd158a067"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-03 20:09:15.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-11-08 22:11:00.000000000","tz":0},"subject":"[manila] Release manila 10.2.0","message":"[manila] Release manila 10.2.0\n\nThis release includes several bugfixes and enhancements committed\nagainst the stable/ussuri branch.\n\nChange-Id: If9fb38c926821ab1a99ae5c6eb2f1503ac34ab06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84085d99b69af8aefd363dba30585e642a4cb08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84085d99b69af8aefd363dba30585e642a4cb08e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
