)]}'
{"id":"openstack%2Freleases~817791","triplet_id":"openstack%2Freleases~master~I1565158578d37548d898ca5d95b3b39cac9c817d","project":"openstack/releases","branch":"master","topic":"ussuri-final","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2021-11-15 13:08:43.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2021-11-12 17:55:21.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"I1565158578d37548d898ca5d95b3b39cac9c817d","subject":"Final ussuri releases for the Cinder project","status":"MERGED","created":"2021-11-12 16:09:39.000000000","updated":"2021-11-15 13:08:43.000000000","submitted":"2021-11-15 13:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817791-ussuri-final","meta_rev_id":"d9934b72ed527eca812bce19ab9c000a2b955f49","_number":817791,"virtual_id_number":817791,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 13:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-11-15 12:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-11-15 11:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-11-12 17:55:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-11-15 12:51:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-11-15 13:08:43.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 16:11:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 17:05:02.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-11-15 11:51:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-15 12:51:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e9ef60fb57185944e9f89c80a510b8c144d947c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-12 16:09:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fb66b239392d2dafff2cb5ad4abb2ca3a2a6c78","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 16:11:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"747a367f414c52fd1119203cdcd8ed9211a595f1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-12 16:11:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f36248bf3c56b92b7455c1c25ef0338951e3aa46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-12 16:17:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8e1b94120b4c890453a7412ef1918c1985752b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 16:50:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b063e44c6db4021b9e6a5456a9a83b5 : SUCCESS in 11m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03be1ed13a334bed8452731e7ddbbdef : FAILURE in 24m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/742a1b78552747be93ad612ab3a2f249 : FAILURE in 9m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5821b11e5ae062ce55caa9d79fdd2e8cf12b2ffa","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-12 17:05:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a627bf1d7bd9257f633e505c15391cd7a2258024","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-12 17:24:37.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b4e8a81bde0f7309fe35f707ab84eb241d6ae91a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-12 17:55:21.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60812e84607d7b3fec837731dd26cfed0f61bb64","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-12 19:03:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"219bc59e0416b71f2d347e0e014790bc3c1722bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 19:32:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5aa4027be28402badffe7a2055f073e : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b94b1f4232174f45ba2b0992b6bcdd3b : SUCCESS in 23m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3665fbe501af46128e73ed9156f36ec8 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cb0e17945525ef0508a9b9f6b9741405f26de959","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-15 11:51:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d6cd9d99f5233bec3fa289e91724b75790340129","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-15 12:51:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d88697d4e88c0cb2d74aeb8d3984bf4190894be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 12:51:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8baa5d46df658e63552f31864c4df36232c060e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:08:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590f0dc19edf44c9a6815a47a9aef502 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19bb2eb20af43419bd2f7b43b3aa410 : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b99a9ea1c9744bb5ab9452eae5db5f17 : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f021927f4d8dfb1c6001c4cb0916b79b8f42955d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:08:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce9ead97c4b8c198fda93e96c7c673d8989118fd","revisions":{"66ff2adae64747c904d7dbd2ac827a9efee5fc99":{"kind":"REWORK","_number":1,"created":"2021-11-12 16:09:39.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/91/817791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/817791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/817791/1"}}},"commit":{"parents":[{"commit":"1f89f773e6f6d21ea20574223418d53b8509a679","subject":"Release Heat Xena RC2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1f89f773e6f6d21ea20574223418d53b8509a679"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-12 16:01:44.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-12 16:01:44.000000000","tz":-300},"subject":"Final ussuri releases for the Cinder project","message":"Final ussuri releases for the Cinder project\n\nThe stable/ussuri branch is soon to enter Extended Maintenance\nphase [0], so here are the final releases in the ussuri series:\n- cinder: release 16.4.2\n- os-brick: release 3.0.8\n- python-cinderclient: release 7.0.2\n\nThe other Cinder deliverables do not require a new release.\nFor the record, their final ussuri series versions are:\n- cinderlib: 2.1.0\n- python-brick-cinderclient-ext: 1.0.2\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2021-October/025276.html\n\nChange-Id: I1565158578d37548d898ca5d95b3b39cac9c817d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66ff2adae64747c904d7dbd2ac827a9efee5fc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66ff2adae64747c904d7dbd2ac827a9efee5fc99"}]},"branch":"refs/heads/master"},"ce9ead97c4b8c198fda93e96c7c673d8989118fd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-12 16:17:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/91/817791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/817791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/817791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/817791/2"}}},"commit":{"parents":[{"commit":"0e01c2b149aad8cc8e1d13fa39e512d90860a0db","subject":"Merge \"List cycle\u0027s releases with process_auto_release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e01c2b149aad8cc8e1d13fa39e512d90860a0db"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-12 16:01:44.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-12 16:17:11.000000000","tz":-300},"subject":"Final ussuri releases for the Cinder project","message":"Final ussuri releases for the Cinder project\n\nThe stable/ussuri branch is soon to enter Extended Maintenance\nphase [0], so here are the final releases in the ussuri series:\n- cinder: release 16.4.2\n- os-brick: release 3.0.8\n- python-cinderclient: release 7.0.2\n\nThe other Cinder deliverables do not require a new release.\nFor the record, their final ussuri series versions are:\n- cinderlib: 2.1.0\n- python-brick-cinderclient-ext: 1.0.2\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2021-October/025276.html\n\nChange-Id: I1565158578d37548d898ca5d95b3b39cac9c817d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce9ead97c4b8c198fda93e96c7c673d8989118fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce9ead97c4b8c198fda93e96c7c673d8989118fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
