)]}'
{"id":"openstack%2Freleases~818404","triplet_id":"openstack%2Freleases~master~Ia7a220d88c74851c9bd49f2e513433de21f917ba","project":"openstack/releases","branch":"master","topic":"yoga-milestone-1","hashtags":[],"change_id":"Ia7a220d88c74851c9bd49f2e513433de21f917ba","subject":"Release blazar-nova for Yoga-1 milestone","status":"ABANDONED","created":"2021-11-18 15:38:36.000000000","updated":"2021-11-18 20:37:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b23f84ca46d95b2694a47d3bc0650a20f3282a9","_number":818404,"virtual_id_number":818404,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-18 15:57:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 15:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-18 15:57:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47a74f6ade119e5e3203de3f2337f9224a9419a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-18 15:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f08a443f56c63e3e79445bdcc6b051ee6d43eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 15:57:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8080ff166d48beaa572b16c8858fe9 : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c44e96f2df3457ab22fcc4495df7e4a : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c3ff356357904217bfbcfe16cf58ba78 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6b23f84ca46d95b2694a47d3bc0650a20f3282a9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-18 20:37:30.000000000","message":"Abandoned\n\nLet\u0027s abandon this as this release would only contain unrelevant changes:\n\ncf659a4 2021-09-23 14:52:05 +0000 Add Python3 yoga unit tests\n57d76b0 2021-09-23 12:26:25 +0000 Adjust lower constraints","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5340686a7519400f85e703ba9be2f5888a0ce638","revisions":{"5340686a7519400f85e703ba9be2f5888a0ce638":{"kind":"REWORK","_number":1,"created":"2021-11-18 15:38:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/04/818404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/818404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/818404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/818404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/818404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/818404/1"}}},"commit":{"parents":[{"commit":"c7a5d2a7a61462697df4cdbb20b44f9e297ab2ab","subject":"Merge \"[yoga] release oslo projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7a5d2a7a61462697df4cdbb20b44f9e297ab2ab"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-11-18 15:38:32.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-11-18 15:38:32.000000000","tz":60},"subject":"Release blazar-nova for Yoga-1 milestone","message":"Release blazar-nova for Yoga-1 milestone\n\nThis is a library release for blazar-nova for the Yoga-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia7a220d88c74851c9bd49f2e513433de21f917ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5340686a7519400f85e703ba9be2f5888a0ce638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5340686a7519400f85e703ba9be2f5888a0ce638"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
