)]}'
{"id":"openstack%2Freleases~819065","triplet_id":"openstack%2Freleases~master~I5357f3a86d740333b2cdbb37cec07b028408f92a","project":"openstack/releases","branch":"master","topic":"mas","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-11-24 15:23:24.000000000","reason":"Elod Illes replied on the change"},"30077":{"account":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"last_update":"2021-11-25 00:55:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5357f3a86d740333b2cdbb37cec07b028408f92a","subject":"sahara release for ussuri","status":"ABANDONED","created":"2021-11-24 08:24:39.000000000","updated":"2021-11-25 00:55:51.000000000","total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"3375f54b4b9475f4a87db92069c4c8bcc880d7c9","_number":819065,"virtual_id_number":819065,"owner":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-24 09:24:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-1,"date":"2021-11-24 15:23:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-11-24 09:14:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 08:25:24.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-11-24 08:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-24 08:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76a06c3c9480c013495db1e629bdd2cb7540b25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-11-24 08:24:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4448535128fd364aec5819699b6dcc83815eef21","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-11-24 08:25:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9ca494c9c46e20e97c39c1d74b185cfc594cbea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 08:37:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48765d30c4046c4a3b78bbb89b67a74 : FAILURE in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0625a3c5b94c44f8b93e09cdfaf4c850 : FAILURE in 4m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/da00778173af4d05966241de31bfa7b4 : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fba03e2e2cd7ca345ab213add553495cdd6e2e70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-24 09:05:14.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bd53b05dfaf7094b4f167e27b22a83025096101f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-11-24 09:13:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"18df4fe393dfbcc9ebfaed5c5c4b85d3f15f5273","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 09:14:52.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c5e4b659fa6fc673665be4ef3bc7d0fefd747366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 09:24:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cd990caf6343da86abcc0e9aff03f6 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4c004b002114e07a3ece6f537313077 : SUCCESS in 8m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/01eb28e3aa814307810e6dc0ac3880b9 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"fd56b040c4de92f25767b6a187f55697e106cf37","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-11-24 10:27:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89ed527e4855cc45b67ad5408530a84747a53970","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-24 15:23:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3375f54b4b9475f4a87db92069c4c8bcc880d7c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-11-25 00:55:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b695a92585786a9454d75be12e94424680544f83","revisions":{"87a45ec189493ed3f926fc8cbe54beb60c7c4560":{"kind":"REWORK","_number":1,"created":"2021-11-24 08:24:39.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/65/819065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/819065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/819065/1"}}},"commit":{"parents":[{"commit":"d7d3b072c2c5b96e3154888bbbbd38b6cb810638","subject":"Merge \"Release os-vif for Yoga-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7d3b072c2c5b96e3154888bbbbd38b6cb810638"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-11-24 08:22:49.000000000","tz":480},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-11-24 08:23:03.000000000","tz":480},"subject":"sahara release for ussuri","message":"sahara release for ussuri\n\nsahara\ngit log --oneline --no-merges 12.0.0 ..69974\n699744e8 (origin/stable/ussuri) Preinstall requirements for docs build and fix lower constraints\n57fd4a11 (gerrit/stable/ussuri) zuul: remove tripleo jobs\n1bfa7c1e Use flask\u0027s builtin methods for test context\n5cacd657 Native zuul v3 grenade job, cleanup -py3\na6ee5223 (tag: 12.0.0.0rc2, tag: 12.0.0) Monkey patch original current_thread _active\n3db9f750 Imported Translations from Zanata\na34a3fbc Update TOX_CONSTRAINTS_FILE for stable/ussuri\n9426b910 Update .gitreview for stable/ussuri\n\nChange-Id: I5357f3a86d740333b2cdbb37cec07b028408f92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87a45ec189493ed3f926fc8cbe54beb60c7c4560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87a45ec189493ed3f926fc8cbe54beb60c7c4560"}]},"branch":"refs/heads/master"},"b695a92585786a9454d75be12e94424680544f83":{"kind":"REWORK","_number":2,"created":"2021-11-24 09:13:12.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/65/819065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/819065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/819065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/819065/2"}}},"commit":{"parents":[{"commit":"d7d3b072c2c5b96e3154888bbbbd38b6cb810638","subject":"Merge \"Release os-vif for Yoga-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7d3b072c2c5b96e3154888bbbbd38b6cb810638"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-11-24 08:22:49.000000000","tz":480},"committer":{"name":"Qiu Fossen","email":"qiujunting@inspur.com","date":"2021-11-24 09:13:12.000000000","tz":0},"subject":"sahara release for ussuri","message":"sahara release for ussuri\n\nsahara\ngit log --oneline --no-merges 12.0.0 ..69974\n699744e8 (origin/stable/ussuri) Preinstall requirements for docs build and fix lower constraints\n57fd4a11 (gerrit/stable/ussuri) zuul: remove tripleo jobs\n1bfa7c1e Use flask\u0027s builtin methods for test context\n5cacd657 Native zuul v3 grenade job, cleanup -py3\na6ee5223 (tag: 12.0.0.0rc2, tag: 12.0.0) Monkey patch original current_thread _active\n3db9f750 Imported Translations from Zanata\na34a3fbc Update TOX_CONSTRAINTS_FILE for stable/ussuri\n9426b910 Update .gitreview for stable/ussuri\n\nChange-Id: I5357f3a86d740333b2cdbb37cec07b028408f92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b695a92585786a9454d75be12e94424680544f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b695a92585786a9454d75be12e94424680544f83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
