)]}'
{"id":"openstack%2Freleases~820244","triplet_id":"openstack%2Freleases~master~I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e","project":"openstack/releases","branch":"master","topic":"ussuri-em-missed-fix","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-03-08 08:28:58.000000000","reason":"Marios Andreou replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-10 14:41:18.000000000","reason":"Change was submitted"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-03-10 14:41:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e","subject":"Release final TripleO stable/ussuri","status":"MERGED","created":"2021-12-02 18:41:03.000000000","updated":"2022-03-10 14:41:18.000000000","submitted":"2022-03-10 14:41:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"820244-ussuri-em-missed-fix","meta_rev_id":"3683dbee33b541444eb0879f57d665cf445cc81b","_number":820244,"virtual_id_number":820244,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 14:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-03-10 14:41:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2022-03-10 14:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2022-03-10 14:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2022-03-10 14:41:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 18:59:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 08:30:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-12-07 14:11:14.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2021-12-09 06:31:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-12-15 14:54:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-03-09 17:09:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"3b48ef557441f2861c9184ef3f1b6ea190526a54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-02 18:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1777abaf75c9127bade829f8a500aa390eb763a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 18:59:57.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"953c1589d834beec8a99b2036f4d58d61c05b811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:24:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27f081d962b40a49d4b2d2a0e72a2ba : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/624050bdc3c148d88d0d5faf180dce1d : FAILURE in 24m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ed2762b217e498ea9f08bb23cd49f9a : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c2716ef411b0d541775cf604293ccec7036ab179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-03 12:31:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6180d9e2017926d0454793ff2adc6fef4b44a1f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 12:33:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"bc4a5ac1d7460c036452f64a12e0907791ee6a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 12:51:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ece075d7d0a4d519abb9d3ff784e7d0 : SUCCESS in 7m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32edccca7bb7441195f70110b262d170 : FAILURE in 16m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/82f6f8eb01f142da9352a3b5630da1c0 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a152626918ee87f8a23a847e75c5081e86c3111f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-07 08:30:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b1e798352873f0fafe758e47629097f6c7e8c45","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-07 14:11:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c78306ba32d81371aa671f904f8ce0b067ddaa57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-09 12:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbcaddfdaa1b366eed9d2aa74a9b5198b7ce0c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-10 08:14:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"20bfce811a5e400e918a1bae4648802aeef45bca","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-10 08:14:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a979e5b19dd7ce2172f6cb1bd350a4cb19469b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 08:38:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb34d77598e4146a166b4ae7f06ec2e : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c62d4171b4c48f09db3fbe758937157 : SUCCESS in 20m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4c43af9484744e8c8b9204ead37463b4 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"593190ac08d9352ddb5aa9cfc9f0b1eb1b308fd4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-10 09:55:03.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"221801c8ecaf2bf4319613fd01a066e7c7d58213","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-13 08:17:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bc7560a8d72ba5ae0e1ee06ca5bb05003a647ab0","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-14 16:33:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e3b550c9216871052a00631adaeb1de13cfb9ac5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 16:35:18.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dfa8a404be9d234f32d3eaee374331ce27a2bf10","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-15 14:54:07.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bca1d6289ae6bb7496f929ba038f89f6513b9ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-22 16:28:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b56fea10fa400507114159fcf453b5ad484bcd61","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-22 16:29:30.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5ad5546c3392d8814374b33eb0ab8a6dceeda39","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 16:35:28.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"74fbcaeb1f6fbaf1725a50fd7665969223b89d3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 16:57:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdcb81f6a56141ee8a60fa397e47b9a0 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07e1dfaeb8d8445eb2077aa51e4f01a9 : FAILURE in 20m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4f5ab983cb34452cbc49170c9899e0b4 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":4},{"id":"30150709eea20e9c52105cd5321ef49a865d0817","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-18 10:34:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a834555dc0c46f127c648afc9fabc41ac393853","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-01 07:14:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47a956c1c336d649ed9e466d833719eac329d376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-01 07:16:42.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"1c2bbc08872b388392e263f3082ef64e5bde5826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 07:47:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23465138658405da3649e95dd44d221 : SUCCESS in 8m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82b90718916f42cc8c2db41e3b80a0f5 : FAILURE in 23m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85864e1e16df4003b3dadab13b74fdab : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":5},{"id":"13167e18540e3d3f07476c8d6ee0b43d2a6184ca","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-04 14:39:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5456c2de470f9956abdf4489b7109453fd24cda5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-04 14:47:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fd24c91f4674aa9ae565cefecfb341939f272ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 15:10:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969a5cef4d36433f9703f36a0b80428a : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f10a28fd1f974ed196a62527ffb84921 : FAILURE in 22m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a171ca8ca7c74387a928c513742c237f : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":6},{"id":"d69609b896fe3d61929dd1f47cba6d26dbcb9764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-03-07 17:38:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1cf00b0987213d4c555c862787469abaeccfab08","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-03-07 17:38:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a318b209d16dbbc1cd04f5debd244081fd7b8168","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 17:39:28.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"bbcf61b35b4fc75a4f3ddefd6518859dfab10cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 18:05:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f98d27c5544a79b95fccf36b301916 : SUCCESS in 8m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c89c233aca3e4693bd89ca75a521882b : FAILURE in 26m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/511582e6fc31427aacaed2bdf4dac8d4 : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":7},{"id":"7c2f94bf4b1e33efcb3da2cd11c85d052e764aaf","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-08 08:28:58.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c4aa23dfef4fb3c12fd8e2c1514afc59d68a4169","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-09 15:04:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3cb0a65dc82c3878c33974650351023afa076b4f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-09 15:09:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6dc68c667126eacbbdaae9d3bfc8b4b5d647a448","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-09 17:09:19.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83eb3b5da86cd2584146baf46bf03ef0c488fb0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-10 13:12:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"6701c15509c3fed580ef4530e0de0e99033015a5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:13:56.000000000","message":"Patch Set 8: -PTL-Approved","accounts_in_message":[],"_revision_number":8},{"id":"b82b3c102cca2dfa3cde1e5eb3a8b7eb1e5307be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:40:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7266ce47f8494b44948066030a4eec00 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5c991d92bb44bd9b59f92d844ac9708 : SUCCESS in 20m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c7b7b81832064b3e938b4209d6af1620 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":8},{"id":"ebdb369429429c6bc697ecd05da9d3432f649c09","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-10 14:12:02.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e144f366e4aa6130b3b81762caf760804c4de8dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:18:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e6429ab8500f593b95b52c48007c13f4e9557b74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:41:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891b32cfc9ff4f8ba25dfdff7dfe2822 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a64aeb8a8a4a3d959580814638c9fd : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1aa4fe68739d451591c393cb63657b40 : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":8},{"id":"5ad683b9fd7e45ed42369d82f241536855c41db0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:41:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3c9bd5bd1105068a279c698c845b535e5a2f8e6f","revisions":{"5ff27f0337f10031945d4e0fcd170e42f14806e1":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:41:03.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/44/820244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/1"}}},"commit":{"parents":[{"commit":"abbba50721c9a8fad554459914f6a013ba837826","subject":"Merge \"kolla/kolla-ansible: transition Rocky and Stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abbba50721c9a8fad554459914f6a013ba837826"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:38:35.000000000","tz":-300},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ff27f0337f10031945d4e0fcd170e42f14806e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ff27f0337f10031945d4e0fcd170e42f14806e1"}]},"branch":"refs/heads/master"},"8991be4013e916b1514a67cb6cac40bc2d5a98a0":{"kind":"REWORK","_number":2,"created":"2021-12-03 12:31:54.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/44/820244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/2"}}},"commit":{"parents":[{"commit":"abbba50721c9a8fad554459914f6a013ba837826","subject":"Merge \"kolla/kolla-ansible: transition Rocky and Stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abbba50721c9a8fad554459914f6a013ba837826"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-03 12:30:57.000000000","tz":-300},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8991be4013e916b1514a67cb6cac40bc2d5a98a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8991be4013e916b1514a67cb6cac40bc2d5a98a0"}]},"branch":"refs/heads/master"},"cffad63a7b92b636dcfa323b4afc7e71c1f091ab":{"kind":"REWORK","_number":3,"created":"2021-12-10 08:14:06.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/820244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/3"}}},"commit":{"parents":[{"commit":"abbba50721c9a8fad554459914f6a013ba837826","subject":"Merge \"kolla/kolla-ansible: transition Rocky and Stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abbba50721c9a8fad554459914f6a013ba837826"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-12-10 08:13:54.000000000","tz":0},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cffad63a7b92b636dcfa323b4afc7e71c1f091ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cffad63a7b92b636dcfa323b4afc7e71c1f091ab"}]},"branch":"refs/heads/master"},"cfd4da3c05167dbb88cbd4c6ff31fdea6af67c1e":{"kind":"REWORK","_number":4,"created":"2021-12-22 16:28:53.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/44/820244/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/4"}}},"commit":{"parents":[{"commit":"abbba50721c9a8fad554459914f6a013ba837826","subject":"Merge \"kolla/kolla-ansible: transition Rocky and Stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abbba50721c9a8fad554459914f6a013ba837826"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-22 16:27:10.000000000","tz":-300},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfd4da3c05167dbb88cbd4c6ff31fdea6af67c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfd4da3c05167dbb88cbd4c6ff31fdea6af67c1e"}]},"branch":"refs/heads/master"},"ee70287587095b7ae735f584d5c97a3e12f13e0a":{"kind":"REWORK","_number":5,"created":"2022-02-01 07:16:42.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/820244/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/5"}}},"commit":{"parents":[{"commit":"abbba50721c9a8fad554459914f6a013ba837826","subject":"Merge \"kolla/kolla-ansible: transition Rocky and Stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abbba50721c9a8fad554459914f6a013ba837826"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-02-01 07:16:42.000000000","tz":0},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee70287587095b7ae735f584d5c97a3e12f13e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee70287587095b7ae735f584d5c97a3e12f13e0a"}]},"branch":"refs/heads/master"},"4f2d4fe58b290f0a82b566e375df53a969bbc3b2":{"kind":"REWORK","_number":6,"created":"2022-03-04 14:47:12.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/820244/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/6"}}},"commit":{"parents":[{"commit":"ace9a3bc36d9a1ef11c4a6af91046e056d8cad4f","subject":"Merge \"Proposed release schedule for Zed (27w)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ace9a3bc36d9a1ef11c4a6af91046e056d8cad4f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-04 14:47:02.000000000","tz":120},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f2d4fe58b290f0a82b566e375df53a969bbc3b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f2d4fe58b290f0a82b566e375df53a969bbc3b2"}]},"branch":"refs/heads/master"},"57a30db65fe475c86ed71bc67539c3d8039c608f":{"kind":"REWORK","_number":7,"created":"2022-03-07 17:38:01.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/44/820244/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/7"}}},"commit":{"parents":[{"commit":"ace9a3bc36d9a1ef11c4a6af91046e056d8cad4f","subject":"Merge \"Proposed release schedule for Zed (27w)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ace9a3bc36d9a1ef11c4a6af91046e056d8cad4f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2022-03-07 17:37:33.000000000","tz":-300},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release\n(excluding tripleo-validations since it is already marked extended\nmaintenance for ussuri).\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/57a30db65fe475c86ed71bc67539c3d8039c608f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/57a30db65fe475c86ed71bc67539c3d8039c608f"}]},"branch":"refs/heads/master"},"3c9bd5bd1105068a279c698c845b535e5a2f8e6f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-03-10 13:12:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/44/820244/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/820244/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/820244/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/820244/8"}}},"commit":{"parents":[{"commit":"3c39fa8f327cb0141a6a3980209a6c6d5b1cb991","subject":"Temporarily set Ussuri as maintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c39fa8f327cb0141a6a3980209a6c6d5b1cb991"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-02 18:26:53.000000000","tz":-300},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-03-10 12:14:05.000000000","tz":60},"subject":"Release final TripleO stable/ussuri","message":"Release final TripleO stable/ussuri\n\nIn preparation for EM of stable/ussuri, do a final release of\nany TripleO projects which have had commits since the last release\n(excluding tripleo-validations since it is already marked extended\nmaintenance for ussuri).\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nDepends-On: I222d8e83a965f1213c244b878e995ed2d47032a5\nChange-Id: I7c66d9ff20ecb22eb3d1436155ba7758f7a3670e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c9bd5bd1105068a279c698c845b535e5a2f8e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c9bd5bd1105068a279c698c845b535e5a2f8e6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
