)]}'
{"id":"openstack%2Freleases~823394","triplet_id":"openstack%2Freleases~master~I8595ee004fdccf00657a026cf70d9dc0b56179a2","project":"openstack/releases","branch":"master","topic":"validations/1.5.0","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-01-07 16:25:47.000000000","reason":"Elod Illes replied on the change"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-01-10 09:03:44.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-01-07 16:25:47.000000000","reason":"Elod Illes replied on the change"},"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2022-01-07 09:22:47.000000000","reason":"Gael Chamoulaud replied on the change"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-01-10 09:03:44.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-01-07 16:25:47.000000000","reason":"Elod Illes replied on the change"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-01-07 15:31:05.000000000","reason":"James Slagle replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-01-07 16:25:47.000000000","reason":"Elod Illes replied on the change"}},"hashtags":[],"change_id":"I8595ee004fdccf00657a026cf70d9dc0b56179a2","subject":"Push Validation Framework release to 1.5.0","status":"MERGED","created":"2022-01-04 14:08:52.000000000","updated":"2022-01-10 09:03:44.000000000","submitted":"2022-01-10 09:03:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"823394-validations/1.5.0","meta_rev_id":"6440a6d965effbe13ac73f907e9dfb049f88b05b","_number":823394,"virtual_id_number":823394,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 09:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2022-01-07 09:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2022-01-07 16:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-01-07 15:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2022-01-10 08:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2022-01-10 08:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-01-07 15:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-01-04 14:09:47.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2022-01-04 14:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-04 14:38:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 08:15:35.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"76c7a5e779a5533875dc3355bdf39b72189fdfb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-04 14:08:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19d09c9a89a039f0073ff4a7f9ea875b22da8e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 14:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13748db13e143ad97c4f60a7d095994 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7dc142c79994c3a8d70223205aaec1f : SUCCESS in 5m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e69fcafbe5214b739eecc28b5c3f93a7 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a42bb218482b1a92f778f53d659839e8c25a63e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-05 10:32:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fae2c7dc24408679eb2b190bffeaef302f93f2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 10:43:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f9beed27a74ed495ca7f727693b9ec : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c738c2dac8a449d8c84a1167c767d5f : FAILURE in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e915b394d0a74cf2ac28710eb1f4e76d : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"05a2b631c3c5e3a524dff0ea2f8a84ea09d24570","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-05 14:22:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2806832f1cdc0bd81b744e7f0049adc514b6839f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 14:37:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe56a9246e9496d9b73241ae35405fd : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65ce21344a9540e386de297b0c509f3c : SUCCESS in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/47df387e6eb447a9b1277080062638fe : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"760f7aac677b467655495cf3c8dc7ec8849fc451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-06 16:58:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"83d8a70cb9c2a3db34eb04a594e6da8ab5912a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:13:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6401b372fa7c4eb0a5a76ba6f70805f3 : SUCCESS in 8m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c346180f08304892bf45c4277526c4ee : SUCCESS in 6m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/94b197256045440c9497b58e90d30280 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":4},{"id":"7fc8f4fe05f191a05669998eff183a0fb27f22a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-07 07:01:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f0911555e04c3c1363f0afbd5185b6762927e441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 07:10:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfda89c651c041b582cdb3a5f0ee6e7b : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6654bea36644c14a0e995e3d46b8cc9 : SUCCESS in 4m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e5cc47d0e31841f7852c88ba1b33fc17 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":5},{"id":"fb023baec293d9e6f425bbe7140185744361a859","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2022-01-07 09:22:47.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"660514c272443cca39a7981b9464383ace33a2d6","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-01-07 15:31:05.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"138dc78d016b26fc60657b6d86f3379ffb9f0c2d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:32:41.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"4e7eda981d2bf0b7dbada98f4344eeb11b795bbb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-07 16:25:47.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"87dcfd070a49a24c936a998e77daefe676d59aca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-10 08:53:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ba1fcd91ec809604322ad0ed3f25fcca22afef8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 08:54:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c5de7949629345f96a455074f615e91dcb26314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:03:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3df165fc4c40bba814ab127ef1abaf : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57bd2bdcf684ee987eaff6c827ce670 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/417bc2ede52143d0a52212f547cdb936 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":5},{"id":"6440a6d965effbe13ac73f907e9dfb049f88b05b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:03:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dd06f18ef3d1833d148e1fcee37a104d1c3cd665","revisions":{"84a12fb8ae4fc9c4613755a92b29725ade30e59b":{"kind":"REWORK","_number":1,"created":"2022-01-04 14:08:52.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/823394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/823394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/823394/1"}}},"commit":{"parents":[{"commit":"4e52e67d5270e02d85e1db45132ae8a0610adb44","subject":"Merge \"Release Kolla projects on maintained stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e52e67d5270e02d85e1db45132ae8a0610adb44"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"subject":"Push Validation Framework release to 1.5.0","message":"Push Validation Framework release to 1.5.0\n\nChange-Id: I8595ee004fdccf00657a026cf70d9dc0b56179a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84a12fb8ae4fc9c4613755a92b29725ade30e59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84a12fb8ae4fc9c4613755a92b29725ade30e59b"}]},"branch":"refs/heads/master"},"56dfcd99947a49edd9997c20803a1f9006475912":{"kind":"REWORK","_number":2,"created":"2022-01-05 10:32:46.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/823394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/823394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/823394/2"}}},"commit":{"parents":[{"commit":"4e52e67d5270e02d85e1db45132ae8a0610adb44","subject":"Merge \"Release Kolla projects on maintained stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e52e67d5270e02d85e1db45132ae8a0610adb44"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-05 10:30:59.000000000","tz":60},"subject":"Push Validation Framework release to 1.5.0 and create bugfix branch","message":"Push Validation Framework release to 1.5.0 and create bugfix branch\n\n* Bump VF to 1.5.0\n* Create bug fix branch for 1.5.0 version\n\nChange-Id: I8595ee004fdccf00657a026cf70d9dc0b56179a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56dfcd99947a49edd9997c20803a1f9006475912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56dfcd99947a49edd9997c20803a1f9006475912"}]},"branch":"refs/heads/master"},"ae28280d315d4340918b8e7d56d1078126d44b69":{"kind":"REWORK","_number":3,"created":"2022-01-05 14:22:50.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/823394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/823394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/823394/3"}}},"commit":{"parents":[{"commit":"4e52e67d5270e02d85e1db45132ae8a0610adb44","subject":"Merge \"Release Kolla projects on maintained stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e52e67d5270e02d85e1db45132ae8a0610adb44"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-05 14:22:33.000000000","tz":60},"subject":"Push Validation Framework release to 1.5.0","message":"Push Validation Framework release to 1.5.0\n\n* Bump VF to 1.5.0\n\nChange-Id: I8595ee004fdccf00657a026cf70d9dc0b56179a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae28280d315d4340918b8e7d56d1078126d44b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae28280d315d4340918b8e7d56d1078126d44b69"}]},"branch":"refs/heads/master"},"64a07c2d520c84c42db85834ca0ca2113b0b71c8":{"kind":"REWORK","_number":4,"created":"2022-01-06 16:58:13.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/823394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/823394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/823394/4"}}},"commit":{"parents":[{"commit":"4e52e67d5270e02d85e1db45132ae8a0610adb44","subject":"Merge \"Release Kolla projects on maintained stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e52e67d5270e02d85e1db45132ae8a0610adb44"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-06 16:58:04.000000000","tz":60},"subject":"Push Validation Framework release to 1.5.0","message":"Push Validation Framework release to 1.5.0\n\n* Bump VF to 1.5.0\n\nChange-Id: I8595ee004fdccf00657a026cf70d9dc0b56179a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64a07c2d520c84c42db85834ca0ca2113b0b71c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64a07c2d520c84c42db85834ca0ca2113b0b71c8"}]},"branch":"refs/heads/master"},"dd06f18ef3d1833d148e1fcee37a104d1c3cd665":{"kind":"REWORK","_number":5,"created":"2022-01-07 07:01:37.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/823394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/823394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/823394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/823394/5"}}},"commit":{"parents":[{"commit":"4e52e67d5270e02d85e1db45132ae8a0610adb44","subject":"Merge \"Release Kolla projects on maintained stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e52e67d5270e02d85e1db45132ae8a0610adb44"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-04 14:07:58.000000000","tz":60},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-07 07:01:29.000000000","tz":60},"subject":"Push Validation Framework release to 1.5.0","message":"Push Validation Framework release to 1.5.0\n\n* Bump VF to 1.5.0\n\nChange-Id: I8595ee004fdccf00657a026cf70d9dc0b56179a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd06f18ef3d1833d148e1fcee37a104d1c3cd665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd06f18ef3d1833d148e1fcee37a104d1c3cd665"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
