)]}'
{"id":"openstack%2Freleases~823779","triplet_id":"openstack%2Freleases~master~I5da83c078a107c0040ec28cbd303afb1941690ca","project":"openstack/releases","branch":"master","topic":"stable_neutron","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-01-11 17:26:41.000000000","reason":"Elod Illes replied on the change"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-01-11 16:36:39.000000000","reason":"Bernard Cafarelli replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-01-13 11:35:32.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-01-11 15:28:14.000000000","reason":"Elod Illes replied on the change"}},"hashtags":[],"change_id":"I5da83c078a107c0040ec28cbd303afb1941690ca","subject":"New stable releases for Neutron","status":"MERGED","created":"2022-01-07 10:02:52.000000000","updated":"2022-01-13 11:35:32.000000000","submitted":"2022-01-13 11:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"823779-stable_neutron","meta_rev_id":"d5a580b77f98c981d27c94e8dc445c620cfab4d0","_number":823779,"virtual_id_number":823779,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 11:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-01-13 11:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-01-11 17:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-01-11 16:36:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-01-13 11:24:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-01-11 16:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 10:04:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 11:57:43.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2022-01-07 14:01:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-11 10:25:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-01-11 14:17:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-01-11 17:26:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-01-13 11:24:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"5ce941b715af05fd03afaecadd23235ac62e40cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-07 10:02:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8921d81f8c747190532f5dfea1a23a77bba353b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:04:15.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"a4b279cef0b0253df10ebfea4d973ddd9a28d27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 11:09:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fa9d5b7f1a4d94a8586ee9cda7fbda : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4dcf818c343442f859aac1c5ae90e89 : TIMED_OUT in 1h 01m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d6357a7418d84c3894588c7b55918b2f : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"02a276202d102ac861b6f5b143e857fcaa8422ef","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-07 14:01:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"89231dd38dde45011af9204a41e867b76af4e2d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 10:12:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9dbdfadbe4176557c3a8e8335e4ab4ed807f719","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 10:14:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"01f1e6d8e3a424310114fa2134f5b2df3fd17061","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:19:07.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f4d7de597669c10fa24d649569cddb824a655970","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-11 10:25:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7707902fdcc2c81438365eeffbcfc5c181b6f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 10:26:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"784ff76e48c336a1a027127bc2951a74af55728f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-11 10:28:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"412064446ae51d8b77fc2246d99f86c7383eabe9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:33:20.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"eab062cb74d02168acd167bb08271b86b028eeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 11:09:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7335c2f2046a41848daf5c9bc773368b : SUCCESS in 8m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f969e3360c0c42598fde43eee1ce9d67 : SUCCESS in 33m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4d9904c747244d71b0feae46d04a31b4 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":3},{"id":"6c279d6916c6173d72b349a4b7ac5a6293cbc233","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-11 14:17:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"199c60b50cb01bc29daf1631877917434b05deeb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 15:15:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"743f908b3447586362feed9ecc5b08da4d8e2a38","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-11 15:28:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"adeb09f755f9d7de57eaadd1c3016bde18c7deed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 15:43:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52fe7697e70efb7a997f65accc64beaf7057d86b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 15:47:12.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"02eea5f9567f9418b755c92022f9cf6c7bbe5edf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-11 15:51:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8db20f85679cb5da6b0f36aa49a4b2980a3c774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-11 15:52:26.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f05351bdf253ad3d46d0fda4370513210c190a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 16:08:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1566fbcf06374a59a1baf4c4feafc856 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b8a34f4de814276ab34e15d16a1f480 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fb5c36ce258640378201b30d0ac9637a : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":5},{"id":"d2788c381745346f998c76651a4c9f3042dcc58d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 16:21:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"95db5e9b0efc1e4ffcc666ab454ac124e43abd74","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 16:22:13.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"af71e7e3f6ed5e67d54260b4a0ff15e7dfe5a852","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 16:25:14.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"e0edd1e7fb3d1925347c41e10bba1dc4e0a01bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 16:36:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951734b546ff4bc48cacc6fcbc1ae64a : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cb40dce466e417aa62c37ca71ea598a : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d880a9ec3e9458babb13cdfa56def04 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":6},{"id":"d65183ff857526b87e78a64c7e655de21b12576d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-11 16:36:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"658f47d2481cdf6a8b73d4adf586de2f4483dba6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-11 17:26:41.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1710425a27ec3ecc3006a781cb8620ab01515332","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-01-13 11:24:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3a2037a84046357a42ce4ee7c73c4465f916eef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:24:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2ecf47cdb8f0cdcfd1351b16075075a4de6a009e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:35:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b05591e6f446c88bf26ba7f42039b3 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7924651b8fdb4f018defa56acbb26282 : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25616b4207b44c14be404f64800b1ad2 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d5a580b77f98c981d27c94e8dc445c620cfab4d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:35:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c0d4fab7b80612d03764721289dc8330cdb845ef","revisions":{"d4763a04a4711ca8b222c3d79307fe3a94842a40":{"kind":"REWORK","_number":1,"created":"2022-01-07 10:02:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/79/823779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/1"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 10:00:03.000000000","tz":60},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.1.2 ebf1bd3f03ebf61ef8bc64bb152a9979231caa6c\nstable/victoria: 17.2.2 bdf3ba10e1bbc7bf9c9fccf9f402545aeef12dd7\nstable/ussuri: 16.4.3 7fa72530b120aa19518eb56c08d84a5c4015e817\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4763a04a4711ca8b222c3d79307fe3a94842a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4763a04a4711ca8b222c3d79307fe3a94842a40"}]},"branch":"refs/heads/master"},"735d3f574df2be0ddadd00187542ea049c01abb3":{"kind":"REWORK","_number":2,"created":"2022-01-11 10:12:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/79/823779/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/2"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-11 10:12:24.000000000","tz":60},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.1.2 ebf1bd3f03ebf61ef8bc64bb152a9979231caa6c\nstable/victoria: 17.2.2 bdf3ba10e1bbc7bf9c9fccf9f402545aeef12dd7\nstable/ussuri: 16.4.3 7fa72530b120aa19518eb56c08d84a5c4015e817\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/735d3f574df2be0ddadd00187542ea049c01abb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/735d3f574df2be0ddadd00187542ea049c01abb3"}]},"branch":"refs/heads/master"},"1decb39520d0690c23adf06834a3fa57ac31b213":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-11 10:26:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/79/823779/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/3"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-01-11 10:26:47.000000000","tz":0},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.1.2 ebf1bd3f03ebf61ef8bc64bb152a9979231caa6c\nstable/victoria: 17.2.2 cf18c3e2e55a8a2d41b040c25a62279fb87e7457\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1decb39520d0690c23adf06834a3fa57ac31b213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1decb39520d0690c23adf06834a3fa57ac31b213"}]},"branch":"refs/heads/master","description":"Edit commit message"},"929f7c134a0382820ca92e9d5949ac4eab7c7dd4":{"kind":"REWORK","_number":4,"created":"2022-01-11 15:43:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/79/823779/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/4"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-11 15:43:03.000000000","tz":60},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.1.2 ebf1bd3f03ebf61ef8bc64bb152a9979231caa6c\nstable/victoria: 17.2.2 cf18c3e2e55a8a2d41b040c25a62279fb87e7457\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/929f7c134a0382820ca92e9d5949ac4eab7c7dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/929f7c134a0382820ca92e9d5949ac4eab7c7dd4"}]},"branch":"refs/heads/master"},"f077fff148dcd791468ab9937a2cb502d7da4526":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-01-11 15:52:26.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/79/823779/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/5"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2022-01-11 15:52:25.000000000","tz":0},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.2.0 ebf1bd3f03ebf61ef8bc64bb152a9979231caa6c\nstable/victoria: 17.3.0 cf18c3e2e55a8a2d41b040c25a62279fb87e7457\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f077fff148dcd791468ab9937a2cb502d7da4526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f077fff148dcd791468ab9937a2cb502d7da4526"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0d4fab7b80612d03764721289dc8330cdb845ef":{"kind":"REWORK","_number":6,"created":"2022-01-11 16:21:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/79/823779/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/823779/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/823779/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/823779/6"}}},"commit":{"parents":[{"commit":"2b587efc2bbe49d4904580f9ee01b0bdccb300f7","subject":"Merge \"Yoga-2 Release for oslo.cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b587efc2bbe49d4904580f9ee01b0bdccb300f7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-07 09:51:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-01-11 16:20:39.000000000","tz":60},"subject":"New stable releases for Neutron","message":"New stable releases for Neutron\n\nstable/wallaby: 18.2.0 65eca44310f0dde8e9c0f47ce33b6a8b61fe96fa\nstable/victoria: 17.3.0 cf18c3e2e55a8a2d41b040c25a62279fb87e7457\n\nChange-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0d4fab7b80612d03764721289dc8330cdb845ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0d4fab7b80612d03764721289dc8330cdb845ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
