)]}'
{"id":"openstack%2Freleases~824394","triplet_id":"openstack%2Freleases~master~Idb5c78e74dea15180ead55a4752ba93041768f9e","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-07-22 11:58:10.000000000","reason":"Change was abandoned"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-10 08:21:19.000000000","reason":"Marcin Juszkiewicz replied on the change"},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2022-07-22 11:58:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idb5c78e74dea15180ead55a4752ba93041768f9e","subject":"Add venus deliverable file for Yoga","status":"ABANDONED","created":"2022-01-12 11:31:29.000000000","updated":"2022-07-22 11:58:10.000000000","total_comment_count":12,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"4ea4807db4418fd9d0e7360d5fda012ebcfe8c05","_number":824394,"virtual_id_number":824394,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 12:16:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-01-12 12:08:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 12:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 11:18:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2022-01-17 14:29:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-05-09 16:53:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-05-10 08:21:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REMOVED"}],"messages":[{"id":"a0a2e154ccf41b0a927ed9dfedaaa0e98013490b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-01-12 11:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dceef353676a1ecd86750939fa11c0a2dce03f1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 12:08:41.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3f83ee1d9e56b539ebe004e64bea13bd9b6c7e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 12:16:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418a47ddb3f24670aeb0688334cb7d47 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ff0303bd9aa431ba7e808a2cd2979c7 : FAILURE in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/79a74597538945b993caa86822bca62b : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"37d52f275120a000fb10a01be794168818ed3a3d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-01-13 11:18:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b79ae7fb94947d1059a3cbdde748d8b1c9392fb4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-01-13 11:22:08.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"71adc4e240b8ac6d2cfa2fa140315625524f2a75","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-17 14:29:21.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b8f31bada7f9e9551fec1f9398b07e7ba93780dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-09 16:53:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"445789e582b5ae0615b93a1c96a2f1d34428b279","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 19:49:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c41460c70846ab0e9c3506945d2a01641fd0eeb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-10 08:21:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"714f9418d91873ffa613f083b2bb252de5c8aac3","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-10 08:21:25.000000000","message":"Removed cc Marcin Juszkiewicz.","accounts_in_message":[],"_revision_number":1},{"id":"4ea4807db4418fd9d0e7360d5fda012ebcfe8c05","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 11:58:10.000000000","message":"Abandoned\n\nyoga deliverables are closed already, this patch is continued in: https://review.opendev.org/c/openstack/releases/+/850129","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e42fbde2a44fcfe14afbbbcc26c52ad067a3a17","revisions":{"4e42fbde2a44fcfe14afbbbcc26c52ad067a3a17":{"kind":"REWORK","_number":1,"created":"2022-01-12 11:31:29.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/94/824394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/824394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/824394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/824394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/824394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/824394/1"}}},"commit":{"parents":[{"commit":"5f4f6295cd022d4ed0565ef5d05875ea5085b15c","subject":"Merge \"[puppet-ironic] Re-create the final stable/ussuri release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f4f6295cd022d4ed0565ef5d05875ea5085b15c"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-01-12 09:08:13.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-01-12 09:08:13.000000000","tz":480},"subject":"Add venus deliverable file for Yoga","message":"Add venus deliverable file for Yoga\n\nChange-Id: Idb5c78e74dea15180ead55a4752ba93041768f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e42fbde2a44fcfe14afbbbcc26c52ad067a3a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e42fbde2a44fcfe14afbbbcc26c52ad067a3a17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
