)]}'
{"id":"openstack%2Freleases~825328","triplet_id":"openstack%2Freleases~master~Ic126cb7d4a7c3e51ecd0f6e8a7ee8030c56254d2","project":"openstack/releases","branch":"master","topic":"ocata-eol","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-20 16:00:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic126cb7d4a7c3e51ecd0f6e8a7ee8030c56254d2","subject":"[OpenStackAnsible] Transition Ocata to End of Life","status":"ABANDONED","created":"2022-01-19 15:16:11.000000000","updated":"2022-01-20 16:00:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eadb8d5516510cf6dabd3c565c70d90cbf82ab63","_number":825328,"virtual_id_number":825328,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-19 15:30:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 15:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-19 15:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-19 15:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1976fefbc8b9d4f6e4f24f3b9d06d2310436ede9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-19 15:16:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a58920703f375dce6e3092ab162bc8705b26d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 15:30:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d7c0f9251e44cc9e600cff63f80bf7 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b73a9537e514443987de050a3a0a6a3b : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/37b370bd679d40e7b27e8da8c37806c6 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5d6a3377f577bc855563c360ba122edc3c063691","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-19 15:35:47.000000000","message":"Topic set to ocata-eol","accounts_in_message":[],"_revision_number":1},{"id":"eadb8d5516510cf6dabd3c565c70d90cbf82ab63","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-20 16:00:03.000000000","message":"Abandoned\n\nsorry, did not see/remember that we have this patch already: https://review.opendev.org/c/openstack/releases/+/822506\nAbandoning this duplicate.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0379bee1c531ee09c12a359bb690c558ec5519e5","revisions":{"0379bee1c531ee09c12a359bb690c558ec5519e5":{"kind":"REWORK","_number":1,"created":"2022-01-19 15:16:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/28/825328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/825328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/825328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/825328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/825328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/825328/1"}}},"commit":{"parents":[{"commit":"21a035544e8992628692eac013144a3b12ca1c9b","subject":"Merge \"[magnum] [wallaby] Release 12.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21a035544e8992628692eac013144a3b12ca1c9b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-01-19 14:03:58.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-01-19 15:15:32.000000000","tz":60},"subject":"[OpenStackAnsible] Transition Ocata to End of Life","message":"[OpenStackAnsible] Transition Ocata to End of Life\n\nThe roles have been transitioned to EOL already and the final patch\nhas merged [1], so now it\u0027s time to EOL openstack-ansible as well.\n\n[1] I922c7cd05708385f375049bcc0f228370a360f6a\n\nChange-Id: Ic126cb7d4a7c3e51ecd0f6e8a7ee8030c56254d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0379bee1c531ee09c12a359bb690c558ec5519e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0379bee1c531ee09c12a359bb690c558ec5519e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
