)]}'
{"id":"openstack%2Freleases~825807","triplet_id":"openstack%2Freleases~master~Ib56cde6e2abac918af37ba570950c71ac070fe5f","project":"openstack/releases","branch":"master","topic":"victoria-stable","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-02-07 07:34:08.000000000","reason":"Marios Andreou replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-02-07 10:23:10.000000000","reason":"Elod Illes replied on the change"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-04-27 11:46:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib56cde6e2abac918af37ba570950c71ac070fe5f","subject":"Release os-net-config for stable/victoria","status":"ABANDONED","created":"2022-01-21 14:07:00.000000000","updated":"2022-04-27 11:46:37.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"490bdb1c87f82dc6825e1d622bdf6b9e19a39a97","_number":825807,"virtual_id_number":825807,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-07 07:49:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-02-07 07:35:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-02-07 10:23:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 14:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 14:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 14:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-01-21 14:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"3f680ac185300c6e2f38ad93d877d40fa202fe2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-21 14:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66062251d93819006c7a18e53c6cf2f3a4fb7f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789b8884f5974ae187f4e197754909b7 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0460342d9b404131a88770bf8589deec : SUCCESS in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d7c40474f8584bd891de01e6346b19ad : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b4160ae81a6e75f460c5f03d0bcad037b07c9806","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-07 07:34:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d758d8ff01e171f8be7bfa3f48c7a15f9285ef8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-07 07:34:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9465c4f07a9fd12e5974e83131dc65385254ad6c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-07 07:35:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"48575132334a7e724e8c0a968855ec91582d83c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 07:49:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5e2d7a7d104d1e8da0190b9d121a16 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a20d2a39ded400bbbf9f45688393bd7 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/02e81897491d4ab7aa6910dc2a402e60 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d3ad58125ec1fc953d49bed7da77f03caa832070","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-02-07 10:23:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"490bdb1c87f82dc6825e1d622bdf6b9e19a39a97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-27 11:46:37.000000000","message":"Abandoned\n\nA different release patch was merged, this is not needed anymore.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bdd0aea9d9acd365b6e385bcd7cef4bd70fdd82","revisions":{"96894292c037de2b158656be7f7ed287c6dc08bd":{"kind":"REWORK","_number":1,"created":"2022-01-21 14:07:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/825807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/825807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/825807/1"}}},"commit":{"parents":[{"commit":"6476a12b4fad26bd77a7229c11991088a203646d","subject":"Merge \"[OpenStackAnsible] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6476a12b4fad26bd77a7229c11991088a203646d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-01-21 14:06:56.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-01-21 14:06:56.000000000","tz":60},"subject":"Release os-net-config for stable/victoria","message":"Release os-net-config for stable/victoria\n\nThis release picks up new commits to os-net-config since\nthe last release from stable/victoria.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 13.2.1..6694552\n6694552 Adding steering_mode config to sriov_pf object\n3ce1bf1 The default max/min tx rates shall not halt deployment\neb08e87 Add ifcfg-* scripts on boot for Mellanox NIC interface\nac934b3 Fix upper constraints\n3c4c81c Update sriov config service to handle nic partitioned PF\n2bf7534 sriov_config: light refactoring\n9780dfa Updating sriov_config.service to run before driverctl.slice and network-pre\n3298a48 Allow linux tap to use routes\n5e0cb3e allow vhost0 to pass list of members\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib56cde6e2abac918af37ba570950c71ac070fe5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96894292c037de2b158656be7f7ed287c6dc08bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96894292c037de2b158656be7f7ed287c6dc08bd"}]},"branch":"refs/heads/master"},"0bdd0aea9d9acd365b6e385bcd7cef4bd70fdd82":{"kind":"REWORK","_number":2,"created":"2022-02-07 07:34:26.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/07/825807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/825807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/825807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/825807/2"}}},"commit":{"parents":[{"commit":"6476a12b4fad26bd77a7229c11991088a203646d","subject":"Merge \"[OpenStackAnsible] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6476a12b4fad26bd77a7229c11991088a203646d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-01-21 14:06:56.000000000","tz":60},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-02-07 07:34:26.000000000","tz":0},"subject":"Release os-net-config for stable/victoria","message":"Release os-net-config for stable/victoria\n\nThis release picks up new commits to os-net-config since\nthe last release from stable/victoria.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 13.2.1..6694552\n6694552 Adding steering_mode config to sriov_pf object\n3ce1bf1 The default max/min tx rates shall not halt deployment\neb08e87 Add ifcfg-* scripts on boot for Mellanox NIC interface\nac934b3 Fix upper constraints\n3c4c81c Update sriov config service to handle nic partitioned PF\n2bf7534 sriov_config: light refactoring\n9780dfa Updating sriov_config.service to run before driverctl.slice and network-pre\n3298a48 Allow linux tap to use routes\n5e0cb3e allow vhost0 to pass list of members\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib56cde6e2abac918af37ba570950c71ac070fe5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0bdd0aea9d9acd365b6e385bcd7cef4bd70fdd82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0bdd0aea9d9acd365b6e385bcd7cef4bd70fdd82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
