)]}'
{"id":"openstack%2Freleases~825824","triplet_id":"openstack%2Freleases~master~I3ee56d32cd0928fe86c1bf0b4c14f385ab3352d3","project":"openstack/releases","branch":"master","topic":"victoria-stable","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-27 17:50:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3ee56d32cd0928fe86c1bf0b4c14f385ab3352d3","subject":"Release swift for stable/victoria","status":"ABANDONED","created":"2022-01-21 15:32:08.000000000","updated":"2022-04-27 17:50:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd0697da1174a2efefe6cc39b48a41f6cc311fef","_number":825824,"virtual_id_number":825824,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-21 15:45:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 15:34:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 15:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 16:20:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"36df8a80929ec4649ad73c2b7ccbcb1220cc832a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-01-21 15:32:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4512f1b4a2caa5f2fbf3753bda8a32c867ac1e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 15:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44aacc082a448669a30aff4e44ed054 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b8fb117a4c04531a14c902282991910 : SUCCESS in 9m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7775e39c6e842f5ba4dd72efaab61a1 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bd0697da1174a2efefe6cc39b48a41f6cc311fef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-27 17:50:15.000000000","message":"Abandoned\n\nswift is now transition to EM, this patch is not needed anymore.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"755f7a30599d45408ad9c51745d9957392532a92","revisions":{"755f7a30599d45408ad9c51745d9957392532a92":{"kind":"REWORK","_number":1,"created":"2022-01-21 15:32:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/24/825824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/825824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/825824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/825824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/825824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/825824/1"}}},"commit":{"parents":[{"commit":"6476a12b4fad26bd77a7229c11991088a203646d","subject":"Merge \"[OpenStackAnsible] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6476a12b4fad26bd77a7229c11991088a203646d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-01-21 15:32:03.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-01-21 15:32:03.000000000","tz":60},"subject":"Release swift for stable/victoria","message":"Release swift for stable/victoria\n\nThis release picks up new commits to swift since\nthe last release from stable/victoria.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 2.26.0..f53500522\nf53500522 Fix cname_lookup test\nc0bb5b2f3 Dep\u0027s should be restricted by upper-constraints\n92b79a4a3 symlink: Ensure headers are returned as a list on py3\na8c2b46da tempurl: Fix PUT upload to temp url on py3\n1032b45e1 Avoid loops when gathering container listings from shards\n2558be42c Populate shrinking shards with shard ranges learnt from root\nf7f0d62c9 Use swiftclient Connections in sharding probe tests\n036f062bf Refactor audit shard tests\n851bdc004 Give probe tests a second chance to pass\n81a4ac4bc Give unit tests a second chance to pass\n68762ebb7 s3-tests: Use my s3compat fork instead\n941ea5953 s3-tests: Only try to copy outputs when there *are* outputs\n1e5107285 Prevent upgrading to pip 21+\nacb742ac1 py3: Fix status comparison in ECGetResponseCollection\nbe74f2141 Memcached client TLS support\n8e744ba24 Pin Bandit on py2\n508c44469 py3: Fix swift-dispersion-populate\nd6a75783b Fixing S3 credential caching region support\nd80b2b590 Update TOX_CONSTRAINTS_FILE for stable/victoria\n3095fb313 Update .gitreview for stable/victoria\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3ee56d32cd0928fe86c1bf0b4c14f385ab3352d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/755f7a30599d45408ad9c51745d9957392532a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/755f7a30599d45408ad9c51745d9957392532a92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
