)]}'
{"id":"openstack%2Freleases~830501","triplet_id":"openstack%2Freleases~master~Idfe1286f9dcd0676f67c9cdd89668de85e57580b","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-26 01:06:58.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-10 05:09:03.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Idfe1286f9dcd0676f67c9cdd89668de85e57580b","subject":"Release puppet-ceph 3.2.0","status":"ABANDONED","created":"2022-02-22 16:53:15.000000000","updated":"2022-08-26 01:06:58.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3517f7280ec7505b726b6e28d8e3e977c7a6d410","_number":830501,"virtual_id_number":830501,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-20 03:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-1,"date":"2022-05-09 10:37:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-04-20 03:27:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 16:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:10:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 11:43:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-05-09 10:37:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ef4c727cf102e2b28824ef46f78e9c6ef836ca6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 16:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7552c6607eac35fcfa7caacb8332db85bea80b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:10:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031733febbaa4aa1b2f735d446cccab1 : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6936a8c96d6407f9b1e58e2dd6766a3 : FAILURE in 6m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ac24672ab5b44cd89e47286974bc7ed9 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fb3536d25856216551b7b0fe46a0b677596df250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-24 15:45:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a211e79ee8aed1dd7796ca3c653b0e10ab299662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:56:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f752255d15874e0a91236424b8e8bcb2 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99f9b68a5eab4ede98b3731be8636c6b : SUCCESS in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c5aa99ef09474c509f331c5b8a0ea8aa : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"37bc0de2b9f6728dc27cb74a3ae059b016931f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-20 03:26:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1007952ae6967c21c52834fc0ae8ac08f5e8fed6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 03:27:44.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c867be3165df5b7a867ce2e7a7f303b8e4d402e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 03:38:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcba714ecd3a4c5aaa45031378e8a761 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d8eb00974c749308013aa7ca873075f : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0b51e54cd8c64126825ee2fadd292a91 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"44e4aae449f7e4208b4f10f4ac7a42b52342aed7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-20 11:43:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17016ae80440a10e3d309cdbd2729a584e501413","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 10:37:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14509e9820e8657e96f2d32012d5bfa5ed659917","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-10 05:09:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3517f7280ec7505b726b6e28d8e3e977c7a6d410","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 01:06:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a84bc2da7373948e4a465645b52282ffb18d74fb","revisions":{"747f7f75c6a983db2537c3029c9393271f76e579":{"kind":"REWORK","_number":1,"created":"2022-02-22 16:53:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/830501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/830501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/830501/1"}}},"commit":{"parents":[{"commit":"88f30d079d3b0530c9eb8314e3f929b71ed1e37f","subject":"Merge \"[wallaby] release oslo.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88f30d079d3b0530c9eb8314e3f929b71ed1e37f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 16:51:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 16:51:08.000000000","tz":540},"subject":"Release puppet-ceph 3.2.0","message":"Release puppet-ceph 3.2.0\n\nA new release of puppet-ceph is created, which includes some fixes like\nthe one[1] to use the CentOS 8 Stream repository instead of CentOS 8\nrepository.\n\n[1] f17e540ed39d78408653d900a13d127326beaf92\n\nChange-Id: Idfe1286f9dcd0676f67c9cdd89668de85e57580b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/747f7f75c6a983db2537c3029c9393271f76e579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/747f7f75c6a983db2537c3029c9393271f76e579"}]},"branch":"refs/heads/master"},"32b7e5be5b327eb2a9faf4fc0604c8daa3a4648b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-24 15:45:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/830501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/830501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/830501/2"}}},"commit":{"parents":[{"commit":"d92da7c37fe65cc58e1d637e6077945e920aa6b8","subject":"Merge \"Release bugfix 20.0 for Ironic in Yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d92da7c37fe65cc58e1d637e6077945e920aa6b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 16:51:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-24 15:45:05.000000000","tz":0},"subject":"Release puppet-ceph 3.2.0","message":"Release puppet-ceph 3.2.0\n\nA new release of puppet-ceph is created, which includes some fixes like\nthe one[1] to use the CentOS 8 Stream repository instead of CentOS 8\nrepository.\n\n[1] f17e540ed39d78408653d900a13d127326beaf92\n\nChange-Id: Idfe1286f9dcd0676f67c9cdd89668de85e57580b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32b7e5be5b327eb2a9faf4fc0604c8daa3a4648b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32b7e5be5b327eb2a9faf4fc0604c8daa3a4648b"}]},"branch":"refs/heads/master","description":"Rebase"},"a84bc2da7373948e4a465645b52282ffb18d74fb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-20 03:26:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/830501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/830501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/830501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/830501/3"}}},"commit":{"parents":[{"commit":"55c0e5c7435430fb5a13bcf1740fe0f2cb288b29","subject":"Merge \"List the Zed cycle signing key as current\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55c0e5c7435430fb5a13bcf1740fe0f2cb288b29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 16:51:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-20 03:26:30.000000000","tz":0},"subject":"Release puppet-ceph 3.2.0","message":"Release puppet-ceph 3.2.0\n\nA new release of puppet-ceph is created, which includes some fixes like\nthe one[1] to use the CentOS 8 Stream repository instead of CentOS 8\nrepository.\n\n[1] f17e540ed39d78408653d900a13d127326beaf92\n\nChange-Id: Idfe1286f9dcd0676f67c9cdd89668de85e57580b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a84bc2da7373948e4a465645b52282ffb18d74fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a84bc2da7373948e4a465645b52282ffb18d74fb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
