)]}'
{"id":"openstack%2Freleases~831467","triplet_id":"openstack%2Freleases~master~I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9","project":"openstack/releases","branch":"master","topic":"glance-yoga-rc1","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-03-03 15:38:05.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-03-03 10:23:37.000000000","reason":"Thierry Carrez replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-03 15:24:04.000000000","reason":"Elod Illes replied on the change"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-03-02 14:01:53.000000000","reason":"Abhishek Kekane replied on the change"}},"hashtags":[],"change_id":"I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9","subject":"Glance Yoga RC1","status":"MERGED","created":"2022-03-02 02:56:51.000000000","updated":"2022-03-03 15:38:05.000000000","submitted":"2022-03-03 15:38:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"831467-glance-yoga-rc1","meta_rev_id":"688b501602df04452d3292fa69eca0fb55f4845e","_number":831467,"virtual_id_number":831467,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 15:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-03-03 10:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-03-03 15:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-03-03 15:24:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-03-02 13:39:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 02:59:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 05:23:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-03-02 12:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-03-02 19:42:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-03-02 19:42:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4b9514cce73494849386878a543c860afe12fef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 02:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"445d107e202f873db79dc97fad41ab5a79676fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 02:58:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"57a84b869675795bc9542ea4cfd14ca0c0fc3ba6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:59:01.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"35f5f2c57c7bceafec7e7dc6971df7c70d482569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 02:59:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"18b1e0e5b13d256be87888e6defd111b1e93f970","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 03:02:08.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"eb118faef4048870eeb952790b3ce925076d0642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 03:14:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6bb265fd024b0dad539b56593f9bcf : SUCCESS in 9m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/392410701c6d407189c399297797bafe : FAILURE in 7m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aaffdd392e7b457c9a206c4e0e9c4339 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":3},{"id":"6026e75c65a96f30925f88efc9f73338720f0821","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-02 05:23:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"372548c6cfc87fa068ea3bb1a5142da8f3ef4b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 05:33:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6e35cc75434f50969da753f8854287 : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fa09a8e804e4875ab48408d71479301 : FAILURE in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/99573ef2d6584a88847cc54e35586931 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ff8623c78ed63136c01bbfcd82bbdb11f85c0d87","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-02 06:01:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed532f9fbc5e3496afd094f3ddeec7d717e8e00d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 12:26:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67cc7d4bf8d7c28667573aeff64c302de38ae382","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:27:47.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"dee25628f50a380b0dddcebb98fa23331d96e61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:36:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a5be13992f40dbb8f07f7822559663 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c96f3816ee042f49cfe9d7ec7f86acc : FAILURE in 6m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ce4a2d4ff9df479785193c76577755f8 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4},{"id":"b9eca9e994e7b74071eede2545fc4bf20beca1b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 12:38:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a94627f7a58a6b8c1ff3d70051c612cd0f2eb395","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:40:12.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"351083b3beae8dfb2258aef2932f7269c0506953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:52:09.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a034b988ab480da3f9f235d26bc57a : SUCCESS in 8m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30166bec8dca4c3b9ee50b2a6c57159b : FAILURE in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8eaa362065b247eb8542bf0fb684d035 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f0cacf268df1a515886541f2602f26be6201e5de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 12:56:33.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"93fa214365f781ed69de908237462a2869972519","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:57:56.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"441c33d7151a2e18946f447a421b11ef450bc302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:08:40.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8522de9110454ea2b6ba73171ff377e7 : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/754b9230307e400386045d7c07439eee : FAILURE in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f3f33c5b9b5648d7b22032f5537ea9c9 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":6},{"id":"f89cdf1e1b1f8175e3ccef61b2054951e8542c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-02 13:37:47.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e58f3b5532c69152ba8e71174765511ff8d976bd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:39:05.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"99e8ce327c26b4811f0f346bcbcdc3a037b69a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:51:04.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc6f3c0e65a4530bdb95d432da14f88 : SUCCESS in 8m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dd3a49c3d524ea99919569dcbc7f545 : SUCCESS in 6m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ca0ba88ab8984bffab71a1136f622744 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":7},{"id":"49f9934b535213df2da260a0f07ce4bf7181459a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-02 14:01:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5fe3257d826f8dcfcd0c71cc7d997d8412f872b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-03-03 10:23:37.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b199c09cb7d2d4203b5318803ea5a013c616092d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-03 15:24:04.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2974cec80d3e9e866b3f0949f82032385347e720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:24:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"14750b4e13cf8e50d0ec50ed46bd073d0dbad389","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:38:05.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5193f1683014c9b8aefcf9d66d3df90 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f4bb2d247e4548b872e4c5313bdf7d : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2c443e53dce43b38c2622a5f605f94f : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":7},{"id":"688b501602df04452d3292fa69eca0fb55f4845e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:38:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4e099def4aff54b4d63c9586ebf6bc8b421dfda1","revisions":{"b3109536452cb263ffcfbbfe8c80485cd5b40743":{"kind":"REWORK","_number":1,"created":"2022-03-02 02:56:51.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/1"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3109536452cb263ffcfbbfe8c80485cd5b40743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3109536452cb263ffcfbbfe8c80485cd5b40743"}]},"branch":"refs/heads/master"},"e3a3576daa307bf178a099d71525e1483cad7769":{"kind":"REWORK","_number":2,"created":"2022-03-02 02:58:55.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/2"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:58:55.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e3a3576daa307bf178a099d71525e1483cad7769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e3a3576daa307bf178a099d71525e1483cad7769"}]},"branch":"refs/heads/master"},"d12efb93a4cd452bc2069e92e76d8a88ee59954b":{"kind":"REWORK","_number":3,"created":"2022-03-02 02:59:46.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/3"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:59:46.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d12efb93a4cd452bc2069e92e76d8a88ee59954b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d12efb93a4cd452bc2069e92e76d8a88ee59954b"}]},"branch":"refs/heads/master"},"a673186e3bfc059ca9893be00f4920c3b73f2281":{"kind":"REWORK","_number":4,"created":"2022-03-02 12:26:37.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/4"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 12:26:25.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a673186e3bfc059ca9893be00f4920c3b73f2281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a673186e3bfc059ca9893be00f4920c3b73f2281"}]},"branch":"refs/heads/master"},"650fc9feb6de6b7e05dd99fb1b14996869aa9814":{"kind":"REWORK","_number":5,"created":"2022-03-02 12:38:56.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/5"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 12:38:56.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/650fc9feb6de6b7e05dd99fb1b14996869aa9814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/650fc9feb6de6b7e05dd99fb1b14996869aa9814"}]},"branch":"refs/heads/master"},"c17c0c18fa0fe11294551fa5fb7bfe3aa8346e2c":{"kind":"REWORK","_number":6,"created":"2022-03-02 12:56:33.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/6"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 12:56:33.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c17c0c18fa0fe11294551fa5fb7bfe3aa8346e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c17c0c18fa0fe11294551fa5fb7bfe3aa8346e2c"}]},"branch":"refs/heads/master"},"4e099def4aff54b4d63c9586ebf6bc8b421dfda1":{"kind":"REWORK","_number":7,"created":"2022-03-02 13:37:47.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/67/831467/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/831467/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/831467/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/831467/7"}}},"commit":{"parents":[{"commit":"d08c63d6f43d69b3a1e9882ccfb26ce82820c22a","subject":"Merge \"Add yoga cycle highlights for cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08c63d6f43d69b3a1e9882ccfb26ce82820c22a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 02:55:56.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2022-03-02 13:37:47.000000000","tz":0},"subject":"Glance Yoga RC1","message":"Glance Yoga RC1\n\nTag Yoga RC1 24.0.0.0rc1\nCut stable branch\n\nChange-Id: I6aa5de3a24e0ebb7b3aa5c5d87836a1c91b643d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e099def4aff54b4d63c9586ebf6bc8b421dfda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e099def4aff54b4d63c9586ebf6bc8b421dfda1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
