)]}'
{"id":"openstack%2Freleases~832343","triplet_id":"openstack%2Freleases~master~I179e307fb78ac056cf52f3df087061c7d3ee676e","project":"openstack/releases","branch":"master","topic":"yoga-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-07 22:31:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I179e307fb78ac056cf52f3df087061c7d3ee676e","subject":"Release Yoga RC1 for adjutant-ui","status":"ABANDONED","created":"2022-03-07 22:21:03.000000000","updated":"2022-03-07 22:31:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ce6ba20867843c512589291c52c68158a9cd391","_number":832343,"virtual_id_number":832343,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-07 22:28:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 22:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 22:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e0be3f656656ed596446ccc67a7045e12dd131a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-07 22:21:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfea71180278e71a3926faa0957236ffad673976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 22:28:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfabcf106f474d5c875291311bc14e83 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14a07587092547aabc3ab1dde12cbdd2 : FAILURE in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/436197a32a794e529eb4ba2db097045c : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1ce6ba20867843c512589291c52c68158a9cd391","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-07 22:31:24.000000000","message":"Abandoned\n\nwrong release type","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76c6afda2f49eb6a299ca69cb4426c7703f5b605","revisions":{"76c6afda2f49eb6a299ca69cb4426c7703f5b605":{"kind":"REWORK","_number":1,"created":"2022-03-07 22:21:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/43/832343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/832343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/832343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/832343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/832343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/832343/1"}}},"commit":{"parents":[{"commit":"3586c80c936af58827097f80b94140c67a6bf573","subject":"Merge \"Minor release (1.6.0) for validations-common with dedicated stable branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3586c80c936af58827097f80b94140c67a6bf573"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 22:20:59.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 22:20:59.000000000","tz":60},"subject":"Release Yoga RC1 for adjutant-ui","message":"Release Yoga RC1 for adjutant-ui\n\nThis patch proposes an RC1 release for adjutant-ui to get any merged\nchanges out prior to the RC1 deadline for Yoga going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, March 10th.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I179e307fb78ac056cf52f3df087061c7d3ee676e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76c6afda2f49eb6a299ca69cb4426c7703f5b605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76c6afda2f49eb6a299ca69cb4426c7703f5b605"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
