)]}'
{"id":"openstack%2Freleases~832419","triplet_id":"openstack%2Freleases~master~I2f0af80ad7225c3fcd2955c764afc94a15d742b6","project":"openstack/releases","branch":"master","topic":"yoga-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-11 11:08:29.000000000","reason":"Elod Illes replied on the change"},"30077":{"account":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"last_update":"2022-03-11 11:26:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f0af80ad7225c3fcd2955c764afc94a15d742b6","subject":"Release Yoga RC1 for sahara-image-elements","status":"MERGED","created":"2022-03-07 23:15:28.000000000","updated":"2022-03-11 11:26:33.000000000","submitted":"2022-03-11 11:26:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832419-yoga-rc1-deadline","meta_rev_id":"af530f2dd75f8d4b3ccaa8f3326bfebf0f1d0809","_number":832419,"virtual_id_number":832419,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 11:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-03-11 11:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-03-11 10:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-03-11 11:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 23:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 23:20:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"state":"REVIEWER"},{"updated":"2022-03-07 23:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 10:48:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a3463dd9b7ed340238b5167467e7a4637593b51e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-07 23:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e8ebf1df4ed392712b1472a0523a3cbea3e2278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:24:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fb9a1670794cf7ba14021058060204 : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f04edb6255e46af9fd5225aab7d835d : SUCCESS in 6m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/89ad2526288b4f1c83d47d87751c87e4 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e756f2728b0703ba16b8876ab67859e423120de8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-10 06:34:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66e68630e652e2d1246b0c7695ef39138a0271d9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-11 10:48:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f52d72b37db49d34311b7397ff87513927e5a21c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-11 11:08:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"938bf40b05281f1733bc78d8a677f9af7c802199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:09:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9544b515d9eaca4b926504607dad558215851b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:26:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7b13049f014725a7be85369d26d15f : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26debdd8f0846a7a9535b411b0589b3 : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02270a4ca27144718b8c9a037e00bf91 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":1},{"id":"af530f2dd75f8d4b3ccaa8f3326bfebf0f1d0809","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:26:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f71b81c883fc066bb6040e6abdd663c2635c861e","revisions":{"f71b81c883fc066bb6040e6abdd663c2635c861e":{"kind":"REWORK","_number":1,"created":"2022-03-07 23:15:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/832419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/832419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/832419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/832419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/832419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/832419/1"}}},"commit":{"parents":[{"commit":"3586c80c936af58827097f80b94140c67a6bf573","subject":"Merge \"Minor release (1.6.0) for validations-common with dedicated stable branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3586c80c936af58827097f80b94140c67a6bf573"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 23:15:22.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 23:15:22.000000000","tz":60},"subject":"Release Yoga RC1 for sahara-image-elements","message":"Release Yoga RC1 for sahara-image-elements\n\nThis patch proposes an RC1 release for sahara-image-elements to get any merged\nchanges out prior to the RC1 deadline for Yoga going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, March 10th.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2f0af80ad7225c3fcd2955c764afc94a15d742b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f71b81c883fc066bb6040e6abdd663c2635c861e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f71b81c883fc066bb6040e6abdd663c2635c861e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
